Skip to content

feat: add cwd for projects that not in the root directory - #134

Merged
Andarist merged 5 commits into
changesets:mainfrom
dmregister:main
Jan 23, 2022
Merged

Andarist merged 5 commits into
changesets:mainfrom
dmregister:main

Conversation

@dmregister

Copy link
Copy Markdown
Contributor

Addresses issue #133

@changeset-bot

changeset-bot Bot commented Jan 11, 2022 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 736eb03

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@changesets/action Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@gyurielf

Copy link
Copy Markdown

Hey @dmregister !

Thanks for the PR.
Can you either create an MD file or write into the PR, to let we know how we can use this feature?

Many thanks!
Gy.

@dmregister

Copy link
Copy Markdown
Contributor Author

@gyurielf I've update the README, is there something else that needs to be updated?

@gyurielf

Copy link
Copy Markdown

@Andarist @mitchellhamilton Can you review it ?

Thank you!

Br,
Gy.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants