Skip to content

feat(usageStats): Add missing client discard reason - #17346

Merged
Lms24 merged 1 commit into
masterfrom
lms/feat-prod-docs-usage-stats-backfill-missing-reasons
Apr 15, 2026
Merged

Lms24 merged 1 commit into
masterfrom
lms/feat-prod-docs-usage-stats-backfill-missing-reasons

Conversation

@Lms24

@Lms24 Lms24 commented Apr 15, 2026

Copy link
Copy Markdown
Member

Adds three missing client (SDK) discard reasons to the /product/stats/ page. These reasons were added over the course of last year (no_parent_span is currently being added) and we missed adding them to docs. This PR backfills them.

ref: https://linear.app/getsentry/issue/SDK-1123/add-client-report-outcome-for-dropped-spans-because-of-missing-parent

@vercel

vercel Bot commented Apr 15, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Actions Updated (UTC)
sentry-docs Ready Ready Preview, Comment Apr 15, 2026 8:03am
1 Skipped Deployment
Project Deployment Actions Updated (UTC)
develop-docs Ignored Ignored Preview Apr 15, 2026 8:03am

Request Review

@Lms24 Lms24 self-assigned this Apr 15, 2026
@linear-code

linear-code Bot commented Apr 15, 2026

Copy link
Copy Markdown

@Lms24
Lms24 merged commit c1c1417 into master Apr 15, 2026
18 checks passed
@Lms24
Lms24 deleted the lms/feat-prod-docs-usage-stats-backfill-missing-reasons branch April 15, 2026 08:13
@github-actions github-actions Bot locked and limited conversation to collaborators May 1, 2026
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants