Skip to content
This repository was archived by the owner on Feb 22, 2022. It is now read-only.

Upgrade to manifestival 0.5.0 - #189

Merged
knative-prow-robot merged 1 commit into
knative:masterfrom
jcrossley3:mf-0.5.0
Apr 14, 2020
Merged

knative-prow-robot merged 1 commit into
knative:masterfrom
jcrossley3:mf-0.5.0

Conversation

@jcrossley3

Copy link
Copy Markdown
Contributor

@houshengbo

Copy link
Copy Markdown

/lgtm
/approve

@knative-prow-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: houshengbo, jcrossley3

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow-robot
knative-prow-robot merged commit b048cec into knative:master Apr 14, 2020
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants