Skip to content
This repository was archived by the owner on Feb 22, 2022. It is now read-only.

Adding doc to change the actual broker - #192

Merged
knative-prow-robot merged 1 commit into
knative:masterfrom
matzew:document_defaultBrokerClass
Apr 17, 2020
Merged

knative-prow-robot merged 1 commit into
knative:masterfrom
matzew:document_defaultBrokerClass

Conversation

@matzew

@matzew matzew commented Apr 17, 2020

Copy link
Copy Markdown
Member

Proposed Changes

  • some doc for defaultBrokerClass

Signed-off-by: Matthias Wessendorf <[email protected]>
@matzew

matzew commented Apr 17, 2020

Copy link
Copy Markdown
Member Author

/assign @aliok

@knative-prow-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: matzew

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@mattmoor mattmoor left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Produced via:
prettier --write --prose-wrap=always $(find -name '*.md' | grep -v vendor | grep -v .github | grep -v docs/cmd/)

Comment thread README.md
Comment on lines +52 to +53
By default the `ChannelBasedBroker` broker will be used, but you can change that by providing
a different `defaultBrokerClass` in the custom resource, like:

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Format markdown:

Suggested change
By default the `ChannelBasedBroker` broker will be used, but you can change that by providing
a different `defaultBrokerClass` in the custom resource, like:
By default the `ChannelBasedBroker` broker will be used, but you can change that
by providing a different `defaultBrokerClass` in the custom resource, like:

@aliok

aliok commented Apr 17, 2020

Copy link
Copy Markdown
Member

/lgtm

@knative-prow-robot
knative-prow-robot merged commit 7971e61 into knative:master Apr 17, 2020
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants