Skip to content
This repository was archived by the owner on Feb 22, 2022. It is now read-only.

Add version consistency test - #42

Merged
knative-prow-robot merged 2 commits into
knative:masterfrom
aliok:version-consistency-test
Nov 18, 2019
Merged

knative-prow-robot merged 2 commits into
knative:masterfrom
aliok:version-consistency-test

Conversation

@aliok

@aliok aliok commented Nov 18, 2019

Copy link
Copy Markdown
Member

Fixes #23

@knative-prow-robot

Copy link
Copy Markdown
Contributor

Hi @aliok. Thanks for your PR.

I'm waiting for a knative member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Comment thread version/version_test.go
continue
}
if v != expectedLabelValue {
t.Errorf("Version info in manifest and operator don't match. got: %v, want: %v. Resource GVK: %v, Resource name: %v", v, expectedLabelValue,

@aliok aliok Nov 18, 2019 •

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

If version info doesn't match, it shows this message multiple times, for each resource in the manifest.
Alternatively, we can switch to failing eagerly with Fatalf

@aliok

aliok commented Nov 18, 2019 •

Copy link
Copy Markdown
Member Author

/assign @houshengbo @matzew

@matzew

matzew commented Nov 18, 2019

Copy link
Copy Markdown
Member

/ok-to-test

Comment thread version/version.go
var (
// Version is the version that will be applied to the KnativeServing resource.
Version = "0.10.0"
Version = "0.10.1"

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

nice catch ;-)

@matzew matzew left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@knative-prow-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: aliok, matzew

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@aliok

aliok commented Nov 18, 2019

Copy link
Copy Markdown
Member Author

@matzew need a new /lgtm . Fixed the test config error.

@matzew

matzew commented Nov 18, 2019

Copy link
Copy Markdown
Member

/lgtm

@knative-prow-robot
knative-prow-robot merged commit 458606e into knative:master Nov 18, 2019
@aliok
aliok deleted the version-consistency-test branch November 18, 2019 13:30
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

checks version of deployment matches version go file

3 participants