Skip to content

[gh-pages] Sync owners file from main - #2341

Merged
knative-prow[bot] merged 1 commit into
knative:gh-pagesfrom
dsimansk:pr/sync-gh-pages-owners
Aug 18, 2026
Merged

knative-prow[bot] merged 1 commit into
knative:gh-pagesfrom
dsimansk:pr/sync-gh-pages-owners

Conversation

@dsimansk

@dsimansk dsimansk commented Aug 18, 2026 •

Copy link
Copy Markdown
Contributor

Fixes #

Proposed Changes

  • [gh-pages] Sync owners file from main

Sync up-to-date OWNERS, OWNERS_ALIASES from main into gh-pages. Currently, the files are stale that significantly limits pool of approvers.

@aliok or @matzew you might be the only approvers with powers left. :)

/cc @aliok @matzew @dprotaso @evankanderson

@knative-prow

knative-prow Bot commented Aug 18, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@knative-prow knative-prow Bot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Aug 18, 2026

@matzew matzew left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@knative-prow knative-prow Bot added lgtm Indicates that a PR is ready to be merged. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Aug 18, 2026
@knative-prow
knative-prow Bot merged commit 82e896a into knative:gh-pages Aug 18, 2026
2 checks passed
@dsimansk
dsimansk deleted the pr/sync-gh-pages-owners branch August 18, 2026 10:04
@knative-prow

knative-prow Bot commented Aug 18, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dsimansk, evankanderson, matzew

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [dsimansk,evankanderson,matzew]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants