Skip to content
This repository was archived by the owner on Jun 24, 2020. It is now read-only.

Applying a released version of the operator - #213

Merged
knative-prow-robot merged 1 commit into
knative:masterfrom
matzew:README_tweaks
Nov 12, 2019
Merged

knative-prow-robot merged 1 commit into
knative:masterfrom
matzew:README_tweaks

Conversation

@matzew

@matzew matzew commented Nov 12, 2019

Copy link
Copy Markdown
Member

Proposed Changes

  • describe how to run an actual release, instead of nightly

@googlebot googlebot added the cla: yes Author(s) signed a CLA. label Nov 12, 2019
@matzew

matzew commented Nov 12, 2019

Copy link
Copy Markdown
Member Author

/assign @houshengbo

@houshengbo houshengbo left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Nice.

@houshengbo houshengbo left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@knative-prow-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: houshengbo, matzew

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow-robot
knative-prow-robot merged commit 9cc982d into knative:master Nov 12, 2019
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants