Skip to content
This repository was archived by the owner on Jun 24, 2020. It is now read-only.

Update vendor and bump k8s version to 1.16.4 - #248

Merged
knative-prow-robot merged 1 commit into
knative:masterfrom
ZhiminXiang:bump-1.16
Jan 8, 2020
Merged

knative-prow-robot merged 1 commit into
knative:masterfrom
ZhiminXiang:bump-1.16

Conversation

@ZhiminXiang

Copy link
Copy Markdown

No description provided.

@googlebot googlebot added the cla: yes Author(s) signed a CLA. label Jan 8, 2020
@ZhiminXiang

Copy link
Copy Markdown
Author

/assign @trshafer

@k4leung4

k4leung4 commented Jan 8, 2020

Copy link
Copy Markdown

/lgtm
/approve

@knative-prow-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: k4leung4, ZhiminXiang

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow-robot
knative-prow-robot merged commit bf987d3 into knative:master Jan 8, 2020
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants