Skip to content

build(deps): bump org.junit.jupiter:junit-jupiter from 5.11.4 to 5.12.0 - #3947

Merged
k8s-ci-robot merged 1 commit into
masterfrom
dependabot/maven/org.junit.jupiter-junit-jupiter-5.12.0
Feb 24, 2025
Merged

k8s-ci-robot merged 1 commit into
masterfrom
dependabot/maven/org.junit.jupiter-junit-jupiter-5.12.0

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Feb 24, 2025

Copy link
Copy Markdown
Contributor

Bumps org.junit.jupiter:junit-jupiter from 5.11.4 to 5.12.0.

Release notes

Sourced from org.junit.jupiter:junit-jupiter's releases.

JUnit 5.12.0 = Platform 1.12.0 + Jupiter 5.12.0 + Vintage 5.12.0

See Release Notes.

New Contributors

Full Changelog: junit-team/junit-framework@r5.11.4...r5.12.0

JUnit 5.12.0-RC2 = Platform 1.12.0-RC2 + Jupiter 5.12.0-RC2 + Vintage 5.12.0-RC2

See Release Notes.

Full Changelog: junit-team/junit-framework@r5.12.0-RC1...r5.12.0-RC2

JUnit 5.12.0-RC1 = Platform 1.12.0-RC1 + Jupiter 5.12.0-RC1 + Vintage 5.12.0-RC1

See Release Notes.

Full Changelog: junit-team/junit-framework@r5.12.0-M1...r5.12.0-RC1

JUnit 5.12.0-M1 = Platform 1.12.0-M1 + Jupiter 5.12.0-M1 + Vintage 5.12.0-M1

See Release Notes.

New Contributors

... (truncated)

Commits
  • 2750b04 Release 5.12.0
  • 9615490 Update log suppressions for tests
  • 0a8d65b Remove duplicated reference to SimpleArgumentConverter
  • cafbcdb Update open-test-reporting to 0.2.0
  • ed1773c Fix version references to be 1.x in Platform and 5.x in Jupiter
  • 1d2f982 Update open-test-reporting to 0.2.0-M3
  • 2b6bab5 Fix indentation
  • d1929bb Merge 5.12.0 release notes
  • 9b7a3c4 Back to snapshots for further development
  • a410c92 Release 5.12.0-RC2
  • Additional commits viewable in compare view

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot merge will merge this PR after your CI passes on it
  • @dependabot squash and merge will squash and merge this PR after your CI passes on it
  • @dependabot cancel merge will cancel a previously requested merge and block automerging
  • @dependabot reopen will reopen this PR if it is closed
  • @dependabot close will close this PR and stop Dependabot recreating it. You can achieve the same result by closing it manually
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

Bumps [org.junit.jupiter:junit-jupiter](https://github.com/junit-team/junit5) from 5.11.4 to 5.12.0.
- [Release notes](https://github.com/junit-team/junit5/releases)
- [Commits](junit-team/junit-framework@r5.11.4...r5.12.0)

---
updated-dependencies:
- dependency-name: org.junit.jupiter:junit-jupiter
  dependency-type: direct:production
  update-type: version-update:semver-minor
...

Signed-off-by: dependabot[bot] <[email protected]>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file java Pull requests that update Java code labels Feb 24, 2025
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. labels Feb 24, 2025

@yue9944882 yue9944882 left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Feb 24, 2025
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dependabot[bot], yue9944882

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 24, 2025
@k8s-ci-robot
k8s-ci-robot merged commit 2f6bc96 into master Feb 24, 2025
@dependabot
dependabot Bot deleted the dependabot/maven/org.junit.jupiter-junit-jupiter-5.12.0 branch February 24, 2025 19:16
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. dependencies Pull requests that update a dependency file java Pull requests that update Java code lgtm "Looks good to me", indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants