Skip to content

Legacy: Automated Generate from openapi release-1.34 - #4252

Merged
k8s-ci-robot merged 4 commits into
kubernetes-client:master-java8from
yue9944882:automated-generate-73da359d
Sep 16, 2025
Merged

k8s-ci-robot merged 4 commits into
kubernetes-client:master-java8from
yue9944882:automated-generate-73da359d

Conversation

@yue9944882

Copy link
Copy Markdown
Member

No description provided.

Signed-off-by: Kubernetes Publisher <[email protected]>
Signed-off-by: Min Jin <[email protected]>
Signed-off-by: Kubernetes Publisher <[email protected]>
Signed-off-by: Min Jin <[email protected]>
Signed-off-by: Kubernetes Publisher <[email protected]>
Signed-off-by: Min Jin <[email protected]>
Signed-off-by: Min Jin <[email protected]>
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Sep 14, 2025
@k8s-ci-robot k8s-ci-robot added the size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. label Sep 14, 2025
@yue9944882
yue9944882 changed the base branch from automated-generate-73da359d to master-java8 September 14, 2025 00:21
@yue9944882 yue9944882 closed this Sep 14, 2025
@yue9944882 yue9944882 reopened this Sep 14, 2025
@yue9944882 yue9944882 changed the title Automated Generate from openapi release-1.34 Legacy: Automated Generate from openapi release-1.34 Sep 15, 2025
@brendandburns

Copy link
Copy Markdown
Contributor

/lgtm
/approve

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Sep 16, 2025
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: brendandburns, yue9944882

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [brendandburns,yue9944882]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot
k8s-ci-robot merged commit 267a313 into kubernetes-client:master-java8 Sep 16, 2025
9 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants