Skip to content

Two ci fixes - #16557

Merged
keszybz merged 2 commits into
systemd:masterfrom
keszybz:two-ci-fixes
Jul 23, 2020
Merged

keszybz merged 2 commits into
systemd:masterfrom
keszybz:two-ci-fixes

Conversation

@keszybz

@keszybz keszybz commented Jul 23, 2020

Copy link
Copy Markdown
Member

No description provided.

keszybz added 2 commits July 23, 2020 12:25
semaphoreci was failing with:
Can't exec "tree": No such file or directory at /tmp/autopkgtest-lxc.v9oand4g/downtmp/build.TIm/src/test/udev-test.pl line 1752.

https://semaphoreci.com/systemd/systemd/branches/pull-request-16551/builds/1
github ci was failing with:

meson.build:685:16: ERROR: String '' cannot be converted to int
@keszybz
keszybz requested a review from daandemeyer July 23, 2020 10:26
@keszybz keszybz added this to the v246 milestone Jul 23, 2020
@keszybz keszybz added the priority Stuff that should enter master quickly, since it fixes a major bug, unbreaks CI or stalls other work label Jul 23, 2020
@poettering poettering added the good-to-merge/waiting-for-ci 👍 PR is good to merge, but CI hasn't passed at time of review. Please merge if you see CI has passed label Jul 23, 2020
@keszybz

keszybz commented Jul 23, 2020

Copy link
Copy Markdown
Member Author

I'll merge this without waiting for the full ubuntu CI, since it doesn't use those code paths anyway, and restart the CI on some of the other PRs.

@keszybz keszybz removed the good-to-merge/waiting-for-ci 👍 PR is good to merge, but CI hasn't passed at time of review. Please merge if you see CI has passed label Jul 23, 2020
@keszybz
keszybz merged commit 402e1e6 into systemd:master Jul 23, 2020
@keszybz
keszybz deleted the two-ci-fixes branch July 23, 2020 15:52
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

ci meson priority Stuff that should enter master quickly, since it fixes a major bug, unbreaks CI or stalls other work

Development

Successfully merging this pull request may close these issues.

2 participants