Repository navigation
fix: mounting of alternate filesystem failing when mounting by monitor list - #5643
Merged
Merged
Conversation
1602077
force-pushed
the
fix/mount-alt-fs-by-monitor-list
branch
from
October 9, 2025 09:45
e1eadc3 to
0371bd4
Compare
1602077
marked this pull request as ready for review
October 9, 2025 09:50
Collaborator
|
/test ci/centos/mini-e2e-helm/k8s-1.33 |
nixpanic
approved these changes
Oct 9, 2025
nixpanic
left a comment
Member
There was a problem hiding this comment.
Looks reasonable to me, thanks!
Madhu-1
approved these changes
Oct 9, 2025
Member
|
@Mergifyio rebase |
Fixes failing to mount shares with an alternate filesystem when mounting by monitor list. Signed-off-by: 1602077 <[email protected]>
Contributor
✅ Branch has been successfully rebased |
ceph-csi-bot
force-pushed
the
fix/mount-alt-fs-by-monitor-list
branch
from
October 9, 2025 12:41
0371bd4 to
f041483
Compare
Member
|
@Mergifyio queue |
Contributor
✅ The pull request has been merged automaticallyDetailsThe pull request has been merged automatically at 44a1fd2 |
54 of 78 tasks
Collaborator
|
/test ci/centos/k8s-e2e-external-storage/1.32 |
Collaborator
|
/test ci/centos/k8s-e2e-external-storage/1.34 |
Collaborator
|
/test ci/centos/k8s-e2e-external-storage/1.33 |
Collaborator
|
/test ci/centos/mini-e2e-helm/k8s-1.32 |
Collaborator
|
/test ci/centos/mini-e2e-helm/k8s-1.34 |
Collaborator
|
/test ci/centos/mini-e2e-helm/k8s-1.33 |
Collaborator
|
/test ci/centos/mini-e2e/k8s-1.32 |
Collaborator
|
/test ci/centos/upgrade-tests-cephfs |
Collaborator
|
/test ci/centos/mini-e2e/k8s-1.33 |
Collaborator
|
/test ci/centos/mini-e2e/k8s-1.34 |
Collaborator
|
/test ci/centos/upgrade-tests-rbd |
Collaborator
|
@Mergifyio backport release-v3.15 |
Contributor
✅ Backports have been createdDetails
|
Closed
6 tasks done
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Describe what this PR does
Passes in missing fsName from volumeContext when mounting by monitor list.
Is there anything that requires special attention
Do you have any questions? No
Is the change backward compatible? Yes
Are there concerns around backward compatibility? No
Related issues
Fixes: #5642
Future concerns
None
Checklist:
guidelines in the developer
guide.
Request
notes
updated with breaking and/or notable changes for the next major release.
Show available bot commands
These commands are normally not required, but in case of issues, leave any of
the following bot commands in an otherwise empty comment in this PR:
/retest ci/centos/<job-name>: retest the<job-name>after unrelatedfailure (please report the failure too!)