Skip to content

util: add explicit nil check before errors.Is(err) check - #5705

Merged
mergify[bot] merged 1 commit into
release-v3.15from
mergify/bp/release-v3.15/pr-5692
Oct 27, 2025
Merged

mergify[bot] merged 1 commit into
release-v3.15from
mergify/bp/release-v3.15/pr-5692

Conversation

@nixpanic

Copy link
Copy Markdown
Member

Describe what this PR does

part of #5677

When error is nil, `!errors.Is(err, ErrConfigNotFound)` evaluates true and
`GetControllerPublishSecretRef` function returns error which should not be the case.

Checklist:

  • Commit Message Formatting: Commit titles and messages follow guidelines in the developer guide.
  • Reviewed the developer guide on Submitting a Pull Request
  • Pending release notes updated with breaking and/or notable changes for the next major release.
  • Documentation has been updated, if necessary.
  • Unit tests have been added, if necessary.
  • Integration tests have been added, if necessary.

Depends-on: #5684


Show available bot commands

These commands are normally not required, but in case of issues, leave any of
the following bot commands in an otherwise empty comment in this PR:

  • /retest ci/centos/<job-name>: retest the <job-name> after unrelated
    failure (please report the failure too!)

Signed-off-by: Praveen M <[email protected]>
(cherry picked from commit 0c8013c)
@Madhu-1

Madhu-1 commented Oct 24, 2025

Copy link
Copy Markdown
Collaborator

@Mergifyio queue

@mergify

mergify Bot commented Oct 24, 2025 •

Copy link
Copy Markdown
Contributor

queue

🛑 The pull request has been removed from the queue default

Details

The merge conditions cannot be satisfied due to failing checks.

You can take a look at Queue: Embarked in merge queue check runs for more details about the failure.

@mergify mergify Bot added the queued label Oct 24, 2025
mergify Bot added a commit that referenced this pull request Oct 24, 2025
@mergify mergify Bot added the ok-to-test Label to trigger E2E tests label Oct 24, 2025
@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/k8s-e2e-external-storage/1.33

@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/k8s-e2e-external-storage/1.32

@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/k8s-e2e-external-storage/1.31

@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/upgrade-tests-cephfs

@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/mini-e2e-helm/k8s-1.32

@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/mini-e2e-helm/k8s-1.31

@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/mini-e2e-helm/k8s-1.33

@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/upgrade-tests-rbd

@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/mini-e2e/k8s-1.32

@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/mini-e2e/k8s-1.31

@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/mini-e2e/k8s-1.33

@ceph-csi-bot ceph-csi-bot removed the ok-to-test Label to trigger E2E tests label Oct 24, 2025
@mergify

mergify Bot commented Oct 24, 2025

Copy link
Copy Markdown
Contributor

This pull request has been removed from the queue for the following reason: checks failed.

The merge conditions cannot be satisfied due to failing checks:

You can check the last failing draft PR here: #5706.

You may have to fix your CI before adding the pull request to the queue again.
If you update this pull request, to fix the CI, it will automatically be requeued once the queue conditions match again.
If you think this was a flaky issue instead, you can requeue the pull request, without updating it, by posting a @mergifyio requeue comment.

@mergify mergify Bot added dequeued and removed queued labels Oct 24, 2025
@nixpanic nixpanic added ok-to-test Label to trigger E2E tests and removed dequeued labels Oct 24, 2025
@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/k8s-e2e-external-storage/1.31

@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/mini-e2e-helm/k8s-1.31

@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/mini-e2e-helm/k8s-1.33

@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/k8s-e2e-external-storage/1.32

@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/upgrade-tests-rbd

@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/mini-e2e/k8s-1.31

@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/mini-e2e/k8s-1.33

@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/mini-e2e-helm/k8s-1.32

@ceph-csi-bot

Copy link
Copy Markdown
Collaborator

/test ci/centos/mini-e2e/k8s-1.32

@ceph-csi-bot ceph-csi-bot removed the ok-to-test Label to trigger E2E tests label Oct 24, 2025
@mergify mergify Bot added the dequeued label Oct 24, 2025
@nixpanic

Copy link
Copy Markdown
Member Author

@Mergifyio requeue

@mergify

mergify Bot commented Oct 24, 2025

Copy link
Copy Markdown
Contributor

requeue

✅ The queue state of this pull request has been cleaned. It can be re-embarked automatically

@nixpanic nixpanic removed the dequeued label Oct 24, 2025
mergify Bot added a commit that referenced this pull request Oct 24, 2025
@mergify

mergify Bot commented Oct 24, 2025

Copy link
Copy Markdown
Contributor

This pull request has been removed from the queue for the following reason: checks failed.

The merge conditions cannot be satisfied due to failing checks:

You can check the last failing draft PR here: #5707.

You may have to fix your CI before adding the pull request to the queue again.
If you update this pull request, to fix the CI, it will automatically be requeued once the queue conditions match again.
If you think this was a flaky issue instead, you can requeue the pull request, without updating it, by posting a @mergifyio requeue comment.

@mergify mergify Bot added dequeued and removed queued labels Oct 24, 2025
@iPraveenParihar

Copy link
Copy Markdown
Contributor

@Mergifyio requeue

@mergify

mergify Bot commented Oct 27, 2025

Copy link
Copy Markdown
Contributor

requeue

✅ The queue state of this pull request has been cleaned. It can be re-embarked automatically

@mergify mergify Bot added queued and removed dequeued labels Oct 27, 2025
mergify Bot added a commit that referenced this pull request Oct 27, 2025
@mergify
mergify Bot merged commit 17a79fc into release-v3.15 Oct 27, 2025
70 checks passed
@mergify
mergify Bot deleted the mergify/bp/release-v3.15/pr-5692 branch October 27, 2025 07:06
@mergify mergify Bot removed the queued label Oct 27, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants