Skip to content

Revert "Revert "Better POJO Support for Document Fields" (#140)" - #192

Merged
wilhuff merged 2 commits into
masterfrom
wilhuff/pojo-mapping
Jan 12, 2019
Merged

wilhuff merged 2 commits into
masterfrom
wilhuff/pojo-mapping

Conversation

@wilhuff

@wilhuff wilhuff commented Jan 11, 2019

Copy link
Copy Markdown
Contributor

This reverts commit 6c2e9a5.

This reverts #140.

This re-applies #76.

@google-oss-bot

google-oss-bot commented Jan 12, 2019 •

Copy link
Copy Markdown
Collaborator

@wilhuff: The following test failed, say /retest to rerun them all:

Test name Commit Details Rerun command
smoke-tests-debug 01a5a27 link /test smoke-tests-debug
Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@mikelehen mikelehen assigned wilhuff and unassigned mikelehen Jan 12, 2019
@wilhuff
wilhuff merged commit 4a02ae3 into master Jan 12, 2019
@wilhuff
wilhuff deleted the wilhuff/pojo-mapping branch January 12, 2019 00:55
@firebase firebase locked and limited conversation to collaborators Oct 12, 2019
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

cla: yes Override cla size/L

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants