Skip to content

Revert "Better POJO Support for Document Fields" - #140

Merged
wilhuff merged 1 commit into
masterfrom
wilhuff/shippable
Nov 28, 2018
Merged

wilhuff merged 1 commit into
masterfrom
wilhuff/shippable

Conversation

@wilhuff

@wilhuff wilhuff commented Nov 27, 2018

Copy link
Copy Markdown
Contributor

This reverts commit ddda926.

This makes master shippable again while we wait for API change approvals.

@mikelehen mikelehen left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

DNLGTM

@wilhuff

wilhuff commented Nov 27, 2018

Copy link
Copy Markdown
Contributor Author

/retest

@wilhuff

wilhuff commented Nov 27, 2018

Copy link
Copy Markdown
Contributor Author

/test connected-check

@vkryachko

Copy link
Copy Markdown
Member

/retest

@wilhuff

wilhuff commented Nov 27, 2018

Copy link
Copy Markdown
Contributor Author

/test connected-check

1 similar comment
@wilhuff

wilhuff commented Nov 27, 2018

Copy link
Copy Markdown
Contributor Author

/test connected-check

@mikelehen mikelehen assigned wilhuff and unassigned mikelehen Nov 28, 2018
@wilhuff
wilhuff merged commit 6c2e9a5 into master Nov 28, 2018
@wilhuff
wilhuff deleted the wilhuff/shippable branch January 7, 2019 22:22
wilhuff added a commit that referenced this pull request Jan 11, 2019
wilhuff added a commit that referenced this pull request Jan 12, 2019
@firebase firebase locked and limited conversation to collaborators Oct 13, 2019
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

cla: yes Override cla size/L

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants