Skip to content
This repository was archived by the owner on Feb 22, 2022. It is now read-only.

Remove the obsolete resource in 0.13 - #131

Merged
knative-prow-robot merged 3 commits into
knative:masterfrom
houshengbo:remove-weebhook
Mar 30, 2020
Merged

knative-prow-robot merged 3 commits into
knative:masterfrom
houshengbo:remove-weebhook

Conversation

@houshengbo

Copy link
Copy Markdown

Issue to be fixed

Fixes #

Proposed Changes

Release Note

@knative-prow-robot knative-prow-robot left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@houshengbo: 0 warnings.

Details

In response to this:

Issue to be fixed

Fixes #

Proposed Changes

Release Note

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@knative-prow-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: houshengbo

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@houshengbo
houshengbo force-pushed the remove-weebhook branch 2 times, most recently from 84e4445 to 20f83c0 Compare March 30, 2020 14:21
@houshengbo houshengbo changed the title Remove the obsolete resource legacysinkbindings webhook in 0.13 Remove the obsolete resource in 0.13 Mar 30, 2020
@houshengbo

Copy link
Copy Markdown
Author

/retest

@k4leung4

Copy link
Copy Markdown

/lgtm

@knative-prow-robot
knative-prow-robot merged commit 388f749 into knative:master Mar 30, 2020
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants