Skip to content
This repository was archived by the owner on Feb 22, 2022. It is now read-only.

Avoid name conflict by using deployment+delimiter+container as the key (#159) - #161

Merged
houshengbo merged 1 commit into
knative:release-0.13from
houshengbo:cherry-pick-13
Mar 31, 2020
Merged

houshengbo merged 1 commit into
knative:release-0.13from
houshengbo:cherry-pick-13

Conversation

@houshengbo

@houshengbo houshengbo commented Mar 31, 2020 •

Copy link
Copy Markdown

Cherry-pick of 044d9c4

@knative-prow-robot knative-prow-robot left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@houshengbo: 0 warnings.

Details

In response to this:

Issue to be fixed

Fixes #

Proposed Changes

Release Note

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@knative-prow-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: houshengbo

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-metrics-robot

Copy link
Copy Markdown

The following is the coverage report on the affected files.
Say /test pull-knative-eventing-operator-go-coverage to re-run this coverage report

File Old Coverage New Coverage Delta
pkg/reconciler/knativeeventing/common/images.go 91.5% 89.7% -1.7

overrideImage := registry.Override[containerName]
func getNewImage(registry *eventingv1alpha1.Registry, containerName, deploymentName string) string {
overrideImage := registry.Override[deploymentName+delimiter+containerName]
if overrideImage == "" {

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

please add a test where override for "deploymentName+delimiter+containerName" and "containerName" is specified, and that "deploymentName+delimiter+containerName" wins out, that is to verify that deploymentName takes precedence over just containerName

Copy link
Copy Markdown
Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

That is what exactly the test cases in image_test.go is doing:

OverrideWithDeploymentContainer, OverridePartialWithDeploymentContainer, OverrideWithDeploymentName

They respectively check overriding both of two containers, overriding one container, and using deploymentName+delimiter+containerName only.

@houshengbo
houshengbo merged commit a19394f into knative:release-0.13 Mar 31, 2020
@houshengbo

Copy link
Copy Markdown
Author

Let me merge it first, to catch up the incoming patch release in 30 mins.
For any other concerns, PLZ open issues. Thanks.

Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants