Skip to content
This repository was archived by the owner on Feb 22, 2022. It is now read-only.

Use finalizers to block CR deletion until all resources are deleted - #172

Merged
knative-prow-robot merged 1 commit into
knative:masterfrom
aliok:finalizer
Apr 1, 2020
Merged

knative-prow-robot merged 1 commit into
knative:masterfrom
aliok:finalizer

Conversation

@aliok

@aliok aliok commented Apr 1, 2020 •

Copy link
Copy Markdown
Member

Issue to be fixed

Fixes #171

Proposed Changes

  • Use a finalizer to block deletion of the CR until all resources are deleted, including the cluster scoped ones
  • Copied from Serving Operator

Release Note

NONE

@knative-prow-robot knative-prow-robot left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@aliok: 0 warnings.

Details

In response to this:

Issue to be fixed

Fixes #171

Proposed Changes

  • Use a finalizer to block deletion of the CR until all resources are deleted, including the cluster scoped ones

Release Note

NONE

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@aliok

aliok commented Apr 1, 2020

Copy link
Copy Markdown
Member Author

@aliok

aliok commented Apr 1, 2020

Copy link
Copy Markdown
Member Author

/retest

@matzew

matzew commented Apr 1, 2020

Copy link
Copy Markdown
Member

/lgtm

@houshengbo houshengbo left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/approve
/lgtm

@knative-prow-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: aliok, houshengbo

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow-robot
knative-prow-robot merged commit 581d36d into knative:master Apr 1, 2020
@aliok
aliok deleted the finalizer branch April 1, 2020 15:43
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

No finalizers set to block deletion of the CR until everything in the manifest is deleted

5 participants