Skip to content
This repository was archived by the owner on Feb 22, 2022. It is now read-only.

Add support of configuring ConfigMaps with CR - #197

Merged
houshengbo merged 9 commits into
knative:masterfrom
houshengbo:pick-configmap
Apr 20, 2020
Merged

houshengbo merged 9 commits into
knative:masterfrom
houshengbo:pick-configmap

Conversation

@houshengbo

@houshengbo houshengbo commented Apr 20, 2020 •

Copy link
Copy Markdown

Issue to be fixed

This PR copied everything from serving-operator repository, to implement the Tranform for ConfigMaps.

Proposed Changes

  • This PR added a new field called config in eventing operator CR to support ConfigMap editing.

Release Note

@knative-prow-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: houshengbo

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow-robot knative-prow-robot left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@houshengbo: 2 warnings.

Details

In response to this:

Issue to be fixed

Fixes #

Proposed Changes

Release Note

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Comment thread pkg/reconciler/knativeeventing/common/config_maps.go
Comment thread pkg/reconciler/knativeeventing/common/config_maps.go

@mattmoor mattmoor left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Produced via:
gofmt -s -w $(find -path './vendor' -prune -o -path './third_party' -prune -o -type f -name '*.go' -print)

@mattmoor mattmoor left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Produced via:
gofmt -s -w $(find -path './vendor' -prune -o -path './third_party' -prune -o -type f -name '*.go' -print)

@mattmoor mattmoor left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Produced via:
gofmt -s -w $(find -path './vendor' -prune -o -path './third_party' -prune -o -type f -name '*.go' -print)

@mattmoor mattmoor left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Produced via:
gofmt -s -w $(find -path './vendor' -prune -o -path './third_party' -prune -o -type f -name '*.go' -print)

@mattmoor mattmoor left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Produced via:
gofmt -s -w $(find -path './vendor' -prune -o -path './third_party' -prune -o -type f -name '*.go' -print)

@mattmoor mattmoor left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Produced via:
gofmt -s -w $(find -path './vendor' -prune -o -path './third_party' -prune -o -type f -name '*.go' -print)

Comment on lines +36 to +39
name string
config configMapData
configMap corev1.ConfigMap
expected corev1.ConfigMap

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Format Go code:

Suggested change
name string
config configMapData
configMap corev1.ConfigMap
expected corev1.ConfigMap
name string
config configMapData
configMap corev1.ConfigMap
expected corev1.ConfigMap

@knative-metrics-robot

Copy link
Copy Markdown

The following is the coverage report on the affected files.
Say /test pull-knative-eventing-operator-go-coverage to re-run this coverage report

File Old Coverage New Coverage Delta
pkg/reconciler/knativeeventing/common/config_maps.go Do not exist 92.3%

@houshengbo houshengbo changed the title Add support of configuring ConfigMap with CR Add support of configuring ConfigMaps with CR Apr 20, 2020
@houshengbo
houshengbo merged commit 8afd85c into knative:master Apr 20, 2020
houshengbo pushed a commit to houshengbo/eventing-operator that referenced this pull request Apr 20, 2020
* Add support of configuring ConfigMap with CR

* Use 0.13.3 as the previous release version

* Update pkg/reconciler/knativeeventing/common/config_maps_test.go

Co-Authored-By: Matt Moore <[email protected]>

* Update pkg/reconciler/knativeeventing/common/config_maps_test.go

Co-Authored-By: Matt Moore <[email protected]>

* Update pkg/reconciler/knativeeventing/common/config_maps_test.go

Co-Authored-By: Matt Moore <[email protected]>

* Update pkg/reconciler/knativeeventing/common/config_maps_test.go

Co-Authored-By: Matt Moore <[email protected]>

* Update pkg/reconciler/knativeeventing/common/config_maps_test.go

Co-Authored-By: Matt Moore <[email protected]>

* Update pkg/reconciler/knativeeventing/common/config_maps_test.go

Co-Authored-By: Matt Moore <[email protected]>

* Update pkg/reconciler/knativeeventing/common/config_maps_test.go

Co-Authored-By: Matt Moore <[email protected]>

Co-authored-by: Matt Moore <[email protected]>
houshengbo pushed a commit that referenced this pull request Apr 20, 2020
* Add support of configuring ConfigMap with CR

* Use 0.13.3 as the previous release version

* Update pkg/reconciler/knativeeventing/common/config_maps_test.go

Co-Authored-By: Matt Moore <[email protected]>

* Update pkg/reconciler/knativeeventing/common/config_maps_test.go

Co-Authored-By: Matt Moore <[email protected]>

* Update pkg/reconciler/knativeeventing/common/config_maps_test.go

Co-Authored-By: Matt Moore <[email protected]>

* Update pkg/reconciler/knativeeventing/common/config_maps_test.go

Co-Authored-By: Matt Moore <[email protected]>

* Update pkg/reconciler/knativeeventing/common/config_maps_test.go

Co-Authored-By: Matt Moore <[email protected]>

* Update pkg/reconciler/knativeeventing/common/config_maps_test.go

Co-Authored-By: Matt Moore <[email protected]>

* Update pkg/reconciler/knativeeventing/common/config_maps_test.go

Co-Authored-By: Matt Moore <[email protected]>

Co-authored-by: Matt Moore <[email protected]>

Co-authored-by: Matt Moore <[email protected]>
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants