Repository navigation
[gh-pages] update OWNERS - #2339
knative-prow-robot wants to merge 2 commits into
Conversation
|
There are empty aliases in OWNER_ALIASES, cleanup is advised. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: knative-prow-robot The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Superseded by #2341 /close |
|
@dsimansk: Closed this PR. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
This is an automated cherry-pick of #2292