Skip to content
This repository was archived by the owner on Jun 24, 2020. It is now read-only.

Added reconciler test and updated deps - #232

Merged
knative-prow-robot merged 3 commits into
knative:masterfrom
jihuin:reconciler-test
Nov 21, 2019
Merged

knative-prow-robot merged 3 commits into
knative:masterfrom
jihuin:reconciler-test

Conversation

@jihuin

@jihuin jihuin commented Nov 21, 2019

Copy link
Copy Markdown
Contributor

Proposed Changes

Release Note

NONE

@googlebot googlebot added the cla: yes Author(s) signed a CLA. label Nov 21, 2019

@knative-prow-robot knative-prow-robot left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@jihuin: 0 warnings.

Details

In response to this:

Proposed Changes

Release Note

NONE

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@jihuin

jihuin commented Nov 21, 2019

Copy link
Copy Markdown
Contributor Author

/assign @trshafer @anniefu @garron

Comment thread pkg/reconciler/reconciler_test.go Outdated
"testing"

_ "knative.dev/pkg/client/injection/client/fake"
_ "knative.dev/pkg/client/injection/kube/client/fake"

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

nit: spacing

@knative-metrics-robot

Copy link
Copy Markdown

The following is the coverage report on the affected files.
Say /test pull-knative-serving-operator-go-coverage to re-run this coverage report

File Old Coverage New Coverage Delta
pkg/reconciler/reconciler.go Do not exist 100.0%

@garron

garron commented Nov 21, 2019

Copy link
Copy Markdown
Contributor

/lgtm

@anniefu

anniefu commented Nov 21, 2019 •

Copy link
Copy Markdown
Contributor

It might be worth minimizing the change to only the package needed for this test instead of updating all of them. i.e. "dep ensure -update knative.dev/pkg"

Running ./update-deps.sh to update all of vendor/ should probably be isolated to its own PR to keep updating misc. packages and adding new feature/test code in independent commits that can be reverted without impacting each other.

@jihuin

jihuin commented Nov 21, 2019

Copy link
Copy Markdown
Contributor Author

@anniefu The changes are all about adding fake clients that are required for the test.

@trshafer

Copy link
Copy Markdown
Contributor

/approve

@knative-prow-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: jihuin, trshafer

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow-robot
knative-prow-robot merged commit d94ecf7 into knative:master Nov 21, 2019
@jihuin
jihuin deleted the reconciler-test branch November 25, 2019 18:46
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants