Skip to content
This repository was archived by the owner on Jun 24, 2020. It is now read-only.

Delete manifest resources in a finalizer - #235

Merged
knative-prow-robot merged 1 commit into
knative:masterfrom
jcrossley3:delete-in-finalizer
Dec 3, 2019
Merged

knative-prow-robot merged 1 commit into
knative:masterfrom
jcrossley3:delete-in-finalizer

Conversation

@jcrossley3

@jcrossley3 jcrossley3 commented Nov 26, 2019 •

Copy link
Copy Markdown
Contributor

This should be more predictable and thread-safe, i.e. The Right Way.

Proposed Changes

  • Delete manifest resources in a finalizer instead of after the KnativeServing resource is already gone.

@googlebot googlebot added the cla: yes Author(s) signed a CLA. label Nov 26, 2019

@knative-prow-robot knative-prow-robot left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@jcrossley3: 1 warning.

Details

In response to this:

This should be more predictable and thread-safe, i.e. The Right Way.

Fixes #

Proposed Changes

  • Delete manifest resources in a finalizer instead of after the KnativeServing resource is already gone.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Comment thread test/e2e/knativeservingdeployment_test.go Outdated
This should be more predictable and thread-safe, i.e. The Right Way.

@matzew matzew left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@knative-prow-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: jcrossley3, matzew

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow-robot
knative-prow-robot merged commit 653df25 into knative:master Dec 3, 2019
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants