Skip to content

ci: Add additional latest tag to konflux builds - #60

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
dsimansk:pr/latest-tag
Jul 31, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
dsimansk:pr/latest-tag

Conversation

@dsimansk

Copy link
Copy Markdown

Summary

  • ci: Add additional latest tag to konflux builds

Add addtional-tags param to -push pipeline. It's created in the beginning of the yaml to make it reachable and more prominent for humans to notice.

/cc @maschmid

@openshift-ci
openshift-ci Bot requested a review from maschmid July 31, 2026 09:43
@openshift-ci

openshift-ci Bot commented Jul 31, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dsimansk

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 31, 2026
@maschmid

Copy link
Copy Markdown

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jul 31, 2026
@openshift-ci

openshift-ci Bot commented Jul 31, 2026

Copy link
Copy Markdown

@dsimansk: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit efac8ae into openshift:master Jul 31, 2026
5 checks passed
pmeida pushed a commit to pmeida/faas-console-plugin that referenced this pull request Aug 3, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants