Conversation
Contributor
|
This pull request was exported from Phabricator. Differential Revision: D19784487 |
💊 CircleCI build failures summary and remediationsAs of commit 95dd965: Commit 95dd965 was recently pushed. Waiting for builds... This comment was automatically generated by Dr. CI (expand for details).Follow this link to opt-out of these comments for your Pull Requests.Please report bugs/suggestions on the GitHub issue tracker. This comment has been revised 2 times. |
houseroad
force-pushed
the
export-D19784487
branch
from
February 7, 2020 05:24
858b06f to
0e38453
Compare
Contributor
|
This pull request was exported from Phabricator. Differential Revision: D19784487 |
…0e6a5e (pytorch#33075) Summary: Pull Request resolved: pytorch#33075 Previous import was 65020daafa9183c769938b4512ce543fd5740f8f Included changes: - **[8b3f7e2e](onnx/onnx@8b3f7e2e)**: Update Dropout and BatchNorm to be Training Friendly (pytorch#2568) <Lara Haidar> - **[61f0bbc5](onnx/onnx@61f0bbc5)**: Fix a bug in ScatterND shape inference (pytorch#2577) <Bowen Bao> - **[05bce9cf](onnx/onnx@05bce9cf)**: add utility function to make reference attribute whose name is not the same as the attribute it refers. (pytorch#2583) <Ke Zhang> - **[71181c83](onnx/onnx@71181c83)**: Clarify spec for constant of shape with dim_n = 0 (pytorch#2567) <Negin Raoof> - **[eadba733](onnx/onnx@eadba733)**: Update sigs.md with link to calendar page (pytorch#2579) <Prasanth Pulavarthi> - **[08562f8e](onnx/onnx@08562f8e)**: Update working-groups.md (pytorch#2580) <Prasanth Pulavarthi> - **[0e718913](onnx/onnx@0e718913)**: Fix Slice op's shape inference logic (pytorch#2526) <Hariharan Seshadri> - **[12111410](onnx/onnx@12111410)**: Add missing spaces to Random*Like doc (pytorch#2572) <Takeshi Watanabe> - **[7e6e61d6](onnx/onnx@7e6e61d6)**: Contributing: fix typos (pytorch#2571) <Maher Jendoubi> - **[bbd604ef](onnx/onnx@bbd604ef)**: Add Einsum op (pytorch#2504) <Negin Raoof> - **[fd3ab73a](onnx/onnx@fd3ab73a)**: Clarify split supports zero length splits (pytorch#2544) <Negin Raoof> - **[6dd73774](onnx/onnx@6dd73774)**: Fix circleci build and drop unsupported Windows builds (pytorch#2565) <Wei-Sheng Chin> - **[b3d201a2](onnx/onnx@b3d201a2)**: Fix the formula of intermediate zero calculation for DynamicQuantizeLinear (pytorch#2556) <Yufeng Li> - **[3613eb25](onnx/onnx@3613eb25)**: Add wording to clarify. (pytorch#2555) <Dwayne Robinson> - **[dfa4384c](onnx/onnx@dfa4384c)**: Fix shape inference for Split with split attribute (pytorch#2328) <Shinichiro Hamaji> - **[684fc1bc](onnx/onnx@684fc1bc)**: Keep symbolic dims in Concat with a single input (pytorch#2418) <Shinichiro Hamaji> Test Plan: ci Reviewed By: hl475 Differential Revision: D19784487 fbshipit-source-id: 3f445d88a014e9d07d1522b9457dd467c100abec
houseroad
force-pushed
the
export-D19784487
branch
from
February 7, 2020 05:35
0e38453 to
95dd965
Compare
Contributor
|
This pull request was exported from Phabricator. Differential Revision: D19784487 |
Contributor
|
This pull request has been merged in 674dca0. |
ttumiel
pushed a commit
to ttumiel/pytorch
that referenced
this pull request
Mar 4, 2020
…0e6a5e (pytorch#33075) Summary: Pull Request resolved: pytorch#33075 Previous import was 65020daafa9183c769938b4512ce543fd5740f8f Included changes: - **[8b3f7e2e](onnx/onnx@8b3f7e2e)**: Update Dropout and BatchNorm to be Training Friendly (pytorch#2568) <Lara Haidar> - **[61f0bbc5](onnx/onnx@61f0bbc5)**: Fix a bug in ScatterND shape inference (pytorch#2577) <Bowen Bao> - **[05bce9cf](onnx/onnx@05bce9cf)**: add utility function to make reference attribute whose name is not the same as the attribute it refers. (pytorch#2583) <Ke Zhang> - **[71181c83](onnx/onnx@71181c83)**: Clarify spec for constant of shape with dim_n = 0 (pytorch#2567) <Negin Raoof> - **[eadba733](onnx/onnx@eadba733)**: Update sigs.md with link to calendar page (pytorch#2579) <Prasanth Pulavarthi> - **[08562f8e](onnx/onnx@08562f8e)**: Update working-groups.md (pytorch#2580) <Prasanth Pulavarthi> - **[0e718913](onnx/onnx@0e718913)**: Fix Slice op's shape inference logic (pytorch#2526) <Hariharan Seshadri> - **[12111410](onnx/onnx@12111410)**: Add missing spaces to Random*Like doc (pytorch#2572) <Takeshi Watanabe> - **[7e6e61d6](onnx/onnx@7e6e61d6)**: Contributing: fix typos (pytorch#2571) <Maher Jendoubi> - **[bbd604ef](onnx/onnx@bbd604ef)**: Add Einsum op (pytorch#2504) <Negin Raoof> - **[fd3ab73a](onnx/onnx@fd3ab73a)**: Clarify split supports zero length splits (pytorch#2544) <Negin Raoof> - **[6dd73774](onnx/onnx@6dd73774)**: Fix circleci build and drop unsupported Windows builds (pytorch#2565) <Wei-Sheng Chin> - **[b3d201a2](onnx/onnx@b3d201a2)**: Fix the formula of intermediate zero calculation for DynamicQuantizeLinear (pytorch#2556) <Yufeng Li> - **[3613eb25](onnx/onnx@3613eb25)**: Add wording to clarify. (pytorch#2555) <Dwayne Robinson> - **[dfa4384c](onnx/onnx@dfa4384c)**: Fix shape inference for Split with split attribute (pytorch#2328) <Shinichiro Hamaji> - **[684fc1bc](onnx/onnx@684fc1bc)**: Keep symbolic dims in Concat with a single input (pytorch#2418) <Shinichiro Hamaji> Test Plan: ci Reviewed By: hl475 Differential Revision: D19784487 fbshipit-source-id: 421cdc3394faeff0168853f4ff065fc599ca3967
laurentdupin
pushed a commit
to laurentdupin/pytorch
that referenced
this pull request
Apr 24, 2026
…0e6a5e (pytorch#33075) Summary: Pull Request resolved: pytorch#33075 Previous import was 65020daafa9183c769938b4512ce543fd5740f8f Included changes: - **[8b3f7e2e](onnx/onnx@8b3f7e2e)**: Update Dropout and BatchNorm to be Training Friendly (pytorch#2568) <Lara Haidar> - **[61f0bbc5](onnx/onnx@61f0bbc5)**: Fix a bug in ScatterND shape inference (pytorch#2577) <Bowen Bao> - **[05bce9cf](onnx/onnx@05bce9cf)**: add utility function to make reference attribute whose name is not the same as the attribute it refers. (pytorch#2583) <Ke Zhang> - **[71181c83](onnx/onnx@71181c83)**: Clarify spec for constant of shape with dim_n = 0 (pytorch#2567) <Negin Raoof> - **[eadba733](onnx/onnx@eadba733)**: Update sigs.md with link to calendar page (pytorch#2579) <Prasanth Pulavarthi> - **[08562f8e](onnx/onnx@08562f8e)**: Update working-groups.md (pytorch#2580) <Prasanth Pulavarthi> - **[0e718913](onnx/onnx@0e718913)**: Fix Slice op's shape inference logic (pytorch#2526) <Hariharan Seshadri> - **[12111410](onnx/onnx@12111410)**: Add missing spaces to Random*Like doc (pytorch#2572) <Takeshi Watanabe> - **[7e6e61d6](onnx/onnx@7e6e61d6)**: Contributing: fix typos (pytorch#2571) <Maher Jendoubi> - **[bbd604ef](onnx/onnx@bbd604ef)**: Add Einsum op (pytorch#2504) <Negin Raoof> - **[fd3ab73a](onnx/onnx@fd3ab73a)**: Clarify split supports zero length splits (pytorch#2544) <Negin Raoof> - **[6dd73774](onnx/onnx@6dd73774)**: Fix circleci build and drop unsupported Windows builds (pytorch#2565) <Wei-Sheng Chin> - **[b3d201a2](onnx/onnx@b3d201a2)**: Fix the formula of intermediate zero calculation for DynamicQuantizeLinear (pytorch#2556) <Yufeng Li> - **[3613eb25](onnx/onnx@3613eb25)**: Add wording to clarify. (pytorch#2555) <Dwayne Robinson> - **[dfa4384c](onnx/onnx@dfa4384c)**: Fix shape inference for Split with split attribute (pytorch#2328) <Shinichiro Hamaji> - **[684fc1bc](onnx/onnx@684fc1bc)**: Keep symbolic dims in Concat with a single input (pytorch#2418) <Shinichiro Hamaji> Test Plan: ci Reviewed By: hl475 Differential Revision: D19784487 fbshipit-source-id: 421cdc3394faeff0168853f4ff065fc599ca3967
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Summary:
Previous import was 65020daafa9183c769938b4512ce543fd5740f8f
Included changes:
Test Plan: ci
Differential Revision: D19784487