Repository navigation
Remove path resolution from internal forks plugin - #23255
Merged
Merged
Conversation
|
Comparing: a3bde79...197d453 Critical size changesIncludes critical production bundles, as well as any change greater than 2%:
Significant size changesIncludes any change greater than 0.2%: (No significant changes) |
acdlite
marked this pull request as ready for review
February 9, 2022 00:08
acdlite
force-pushed
the
no-resolve-forks
branch
from
February 9, 2022 00:25
1dbee60 to
d283f4a
Compare
Alternative to #23254 Our build script has a custom plugin to resolve internal module forks. Currently, it uses require.resolve to resolve the path to a real file on disk. Instead, I've updated all the forked module paths to match their location on disk, relative to the project root, to remove the need to resolve them in the build script's runtime. The main motivation is because require.resolve doesn't work with ESM modules, but aside from that, hardcoding the relative paths is more predictable — the Node module resolution algorithm is complicated, and we don't really need its features for this purpose.
acdlite
force-pushed
the
no-resolve-forks
branch
2 times, most recently
from
February 9, 2022 04:34
25a68df to
197d453
Compare
sebmarkbage
approved these changes
Feb 9, 2022
facebook-github-bot
pushed a commit
to react/react-native
that referenced
this pull request
Feb 17, 2022
Summary: This sync includes the following changes: - **[27b5699](react/react@27b569969 )**: Simplify cache pool contexts ([#23280](react/react#23280)) //<Andrew Clark>// - **[1fb0d06](react/react@1fb0d0687 )**: [Devtools][Transition Tracing] Add Transition callbacks to createRoot ([#23276](react/react#23276)) //<Luna Ruan>// - **[a6987be](react/react@a6987bee7 )**: add <TracingMarker> component boilerplate ([#23275](react/react#23275)) //<Luna Ruan>// - **[796fff5](react/react@796fff548 )**: Allow suspending outside a Suspense boundary ([#23267](react/react#23267)) //<Andrew Clark>// - **[64223fe](react/react@64223fed8 )**: Fix: Multiple hydration errors in same render ([#23273](react/react#23273)) //<Andrew Clark>// - **[efd8f64](react/react@efd8f6442 )**: Resolve default onRecoverableError at root init ([#23264](react/react#23264)) //<Andrew Clark>// - **[e0af1aa](react/react@e0af1aabe )**: Fix wrong context argument to `apply` //<Andrew Clark>// - **[9b5e051](react/react@9b5e0517b )**: Remove deprecated wildcard folder mapping ([#23256](react/react#23256)) //<Andrew Clark>// - **[274b9fb](react/react@274b9fb16 )**: Remove path resolution from internal forks plugin ([#23255](react/react#23255)) //<Andrew Clark>// Changelog: [General][Changed] - React Native sync for revisions a3bde79...27b5699 jest_e2e[run_all_tests] Reviewed By: rickhanlonii, kacieb Differential Revision: D34241986 fbshipit-source-id: f6ab62df2a918728864283b4f13201275eb3b8a3
zhengjitf
pushed a commit
to zhengjitf/react
that referenced
this pull request
Apr 15, 2022
Alternative to react#23254 Our build script has a custom plugin to resolve internal module forks. Currently, it uses require.resolve to resolve the path to a real file on disk. Instead, I've updated all the forked module paths to match their location on disk, relative to the project root, to remove the need to resolve them in the build script's runtime. The main motivation is because require.resolve doesn't work with ESM modules, but aside from that, hardcoding the relative paths is more predictable — the Node module resolution algorithm is complicated, and we don't really need its features for this purpose.
Saadnajmi
pushed a commit
to Saadnajmi/react-native-macos
that referenced
this pull request
Jan 15, 2023
Summary: This sync includes the following changes: - **[27b5699](react/react@27b569969 )**: Simplify cache pool contexts ([react#23280](react/react#23280)) //<Andrew Clark>// - **[1fb0d06](react/react@1fb0d0687 )**: [Devtools][Transition Tracing] Add Transition callbacks to createRoot ([react#23276](react/react#23276)) //<Luna Ruan>// - **[a6987be](react/react@a6987bee7 )**: add <TracingMarker> component boilerplate ([react#23275](react/react#23275)) //<Luna Ruan>// - **[796fff5](react/react@796fff548 )**: Allow suspending outside a Suspense boundary ([react#23267](react/react#23267)) //<Andrew Clark>// - **[64223fe](react/react@64223fed8 )**: Fix: Multiple hydration errors in same render ([react#23273](react/react#23273)) //<Andrew Clark>// - **[efd8f64](react/react@efd8f6442 )**: Resolve default onRecoverableError at root init ([react#23264](react/react#23264)) //<Andrew Clark>// - **[e0af1aa](react/react@e0af1aabe )**: Fix wrong context argument to `apply` //<Andrew Clark>// - **[9b5e051](react/react@9b5e0517b )**: Remove deprecated wildcard folder mapping ([react#23256](react/react#23256)) //<Andrew Clark>// - **[274b9fb](react/react@274b9fb16 )**: Remove path resolution from internal forks plugin ([react#23255](react/react#23255)) //<Andrew Clark>// Changelog: [General][Changed] - React Native sync for revisions a3bde79...27b5699 jest_e2e[run_all_tests] Reviewed By: rickhanlonii, kacieb Differential Revision: D34241986 fbshipit-source-id: f6ab62df2a918728864283b4f13201275eb3b8a3
mrizwanashiq
pushed a commit
to mrizwanashiq/react
that referenced
this pull request
Jun 25, 2026
Alternative to react#23254 Our build script has a custom plugin to resolve internal module forks. Currently, it uses require.resolve to resolve the path to a real file on disk. Instead, I've updated all the forked module paths to match their location on disk, relative to the project root, to remove the need to resolve them in the build script's runtime. The main motivation is because require.resolve doesn't work with ESM modules, but aside from that, hardcoding the relative paths is more predictable — the Node module resolution algorithm is complicated, and we don't really need its features for this purpose.
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Alternative to #23254
Our build script has a custom plugin to resolve internal module forks. Currently, it uses require.resolve to resolve the path to a real file on disk.
Instead, I've updated all the forked module paths to match their location on disk, relative to the project root, to remove the need to resolve them in the build script's runtime.
The main motivation is because require.resolve doesn't work with ESM modules, but aside from that, hardcoding the relative paths is more predictable — the Node module resolution algorithm is complicated, and we don't really need its features for this purpose.