Conversation
Performance and stability improvements on all platforms. [email protected] Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng Change-Id: Ie97ba96c54f0cc820ab9700b3318e6826cea5cc1 Reviewed-on: https://chromium-review.googlesource.com/710396 Reviewed-by: v8 autoroll <[email protected]> Cr-Commit-Position: refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
[email protected] Change-Id: I323a4b4e27c3c623cd58aef8a7a13a5dcf135755 Reviewed-on: https://chromium-review.googlesource.com/716116 Reviewed-by: Michael Hablich <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#1} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
[email protected] Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng Change-Id: I419977d23adf80a7ef57e8b3d17307847f48406f Reviewed-on: https://chromium-review.googlesource.com/715779 Reviewed-by: v8 autoroll <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#2} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
BUG=chromium:318206 [email protected], [email protected] LOG=n NOTRY=true NOPRESUBMIT=true Change-Id: I23869d1c27c39e940bf30664e7d4d6f5eb92b58d Reviewed-on: https://chromium-review.googlesource.com/715764 Reviewed-by: Michael Hablich <[email protected]> Commit-Queue: Michael Hablich <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#3} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
[email protected] Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng Change-Id: Ib7b011eeb7ffd1ab55559559fc2399d5ec0fa68c Reviewed-on: https://chromium-review.googlesource.com/716236 Reviewed-by: v8 autoroll <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#4} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
Revision: a445b97 Should fix build problem on 6.3. NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: I8c679afc5003ebe3ccae926e7a752fa1f9e67bd7 Reviewed-on: https://chromium-review.googlesource.com/716225 Reviewed-by: Michael Hablich <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#5} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
[email protected] Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng Change-Id: I3867c48296c3c5d5f636d35e3cfd0376b433c7c1 Reviewed-on: https://chromium-review.googlesource.com/716237 Reviewed-by: v8 autoroll <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#6} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
This reverts commit d9a2584. This is for 6.3 branch (to go with Chromium 63 branch). Reason for revert: I'm reverting this CL for a few reasons. #2 is the most significant and I should have thought of that before making a switch. Sorry for that. 1) perf-regression: http://crbug.com/769706 2) http://crbug.com/612010 : ICU timezone update is not propagated to zygote process so that new tabs will hold on to an old timezone even after a timezone change on Linux and Chrome OS. 3) http://crbug.com/754053 : OS timezone detection issues on macOS 10.13, Ubutu 16, RHEL 7, SuSe Linux 12 or newer. ; it's being fixed. So, it actually ok. 4) http://crbug.com/771868 : timezone wrong in gmail: If it's due to Revision: 28ef8dc BUG=chromium:612010,chromium:769706,chromium:771868,v8:2137,v8:6031,v8:6076 LOG=N NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected], [email protected] Change-Id: Id0093c2bc69e5bc1d3b7147b7bbcc633ed625a45 Reviewed-on: https://chromium-review.googlesource.com/721919 Reviewed-by: Adam Klein <[email protected]> Commit-Queue: Jungshik Shin <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#7} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
[email protected] Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng Change-Id: Ib8890215f805bc52e0934b85d3148004d8905202 Reviewed-on: https://chromium-review.googlesource.com/721980 Reviewed-by: v8 autoroll <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#8} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
Revision: 69bd294 Merge to 6.3 branch - remove unused Runtime_GetLanguageTagVariants - add test for another related bug (chromium:770452) as well as for chromium:770450 . BUG=chromium:770450,chromium:770452 LOG=N NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Cq-Include-Trybots: master.tryserver.v8:v8_linux_noi18n_rel_ng Change-Id: Ia2c6dd2156c51995fb18228fc3062a86e78d719c Reviewed-on: https://chromium-review.googlesource.com/721844 Reviewed-by: Adam Klein <[email protected]> Commit-Queue: Jungshik Shin <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#9} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
[email protected] Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng Change-Id: I0a6e5fbf8480961e228d48df9da3eccc9c1d2b21 Reviewed-on: https://chromium-review.googlesource.com/721981 Reviewed-by: v8 autoroll <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#10} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
…cts. Revision: 6aa9f43 BUG=chromium:763073 LOG=N NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: If8a8ebf9e176e862912010deb1a54ae484bd45e3 Reviewed-on: https://chromium-review.googlesource.com/722639 Reviewed-by: Benedikt Meurer <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#11} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
… bits). Revision: a803fad BUG=v8:6916 LOG=N NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: I4236e37ee9693bed3f4bb65e43b33d8da94cf634 Reviewed-on: https://chromium-review.googlesource.com/722640 Reviewed-by: Benedikt Meurer <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#12} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
[email protected] Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng Change-Id: I7579594e7dafe60bfb2b65df09fb3425d8c04d40 Reviewed-on: https://chromium-review.googlesource.com/722082 Reviewed-by: v8 autoroll <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#13} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
Merged: [ic] Do access checks when storing via JSGlobalProxy. Revision: 5ea95fe Merged: [ic] Fix storing to JSGlobalProxy having JSProxy in prototype chain. Revision: b19a1ba BUG=chromium:764219 LOG=N NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: Icc7dde6ba3faae1aee29c4f3ac1aaa27d94f3124 Reviewed-on: https://chromium-review.googlesource.com/721443 Reviewed-by: Toon Verwaest <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#14} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
[email protected] Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng Change-Id: I11f567e375bacb9c2774b2ecd9efb5792426021c Reviewed-on: https://chromium-review.googlesource.com/721943 Reviewed-by: v8 autoroll <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#15} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
Parser creates a FunctionState for default ctors, which affects the next_function_is_likely_called logic. PreParser needs to match that logic, so that Parser and PreParser agree about which functions are skippable. BUG=v8:5515, chromium:773576 NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true Change-Id: I96cb6f5aa68e74389a863355f70a34693a2d1329 Reviewed-on: https://chromium-review.googlesource.com/712579 Commit-Queue: Marja Hölttä <[email protected]> Reviewed-by: Adam Klein <[email protected]> Cr-Original-Commit-Position: refs/heads/master@{#48511}(cherry picked from commit dd18faf) Reviewed-on: https://chromium-review.googlesource.com/723519 Reviewed-by: Michael Hablich <[email protected]> Commit-Queue: Michael Hablich <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#16} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
This reverts commit 593b0d8. Reason for revert: broke some branch builders like https://build.chromium.org/p/client.v8.branches/builders/V8%20arm%20-%20sim%20-%20beta%20branch%20-%20debug Original change's description: > Merged: Correct the misuse of uloc_{to,from}LanguageTag > > Revision: 69bd294 > > Merge to 6.3 branch > > - remove unused Runtime_GetLanguageTagVariants > - add test for another related bug (chromium:770452) as well as for > chromium:770450 . > > BUG=chromium:770450,chromium:770452 > LOG=N > NOTRY=true > NOPRESUBMIT=true > NOTREECHECKS=true > R=[email protected] > > Cq-Include-Trybots: master.tryserver.v8:v8_linux_noi18n_rel_ng > Change-Id: Ia2c6dd2156c51995fb18228fc3062a86e78d719c > Reviewed-on: https://chromium-review.googlesource.com/721844 > Reviewed-by: Adam Klein <[email protected]> > Commit-Queue: Jungshik Shin <[email protected]> > Cr-Commit-Position: refs/branch-heads/6.3@{#9} > Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} > Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432} [email protected],[email protected],[email protected] Change-Id: I37018f8241efe1431f453ff55cf8216a5daa66de No-Presubmit: true No-Tree-Checks: true No-Try: true Bug: chromium:770450, chromium:770452 Cq-Include-Trybots: master.tryserver.v8:v8_linux_noi18n_rel_ng Reviewed-on: https://chromium-review.googlesource.com/723323 Reviewed-by: Michael Hablich <[email protected]> Commit-Queue: Michael Hablich <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#17} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
This reverts commit d734370. Reason for revert: broke some branch builders like https://build.chromium.org/p/client.v8.branches/builders/V8%20arm%20-%20sim%20-%20beta%20branch%20-%20debug Original change's description: > Merged: Revert "Enable icu-timezone-data by default" > > This reverts commit d9a2584. > This is for 6.3 branch (to go with Chromium 63 branch). > > Reason for revert: > > I'm reverting this CL for a few reasons. #2 is the most significant and > I should have thought of that before making a switch. Sorry for that. > > 1) perf-regression: http://crbug.com/769706 > 2) http://crbug.com/612010 : ICU timezone update is not propagated to > zygote process so that new tabs will hold on to an old timezone even > after a timezone change on Linux and Chrome OS. > > 3) http://crbug.com/754053 : OS timezone detection issues on macOS > 10.13, Ubutu 16, RHEL 7, SuSe Linux 12 or newer. ; it's being fixed. So, > it actually ok. > > 4) http://crbug.com/771868 : timezone wrong in gmail: If it's due to > > Revision: 28ef8dc > > BUG=chromium:612010,chromium:769706,chromium:771868,v8:2137,v8:6031,v8:6076 > LOG=N > NOTRY=true > NOPRESUBMIT=true > NOTREECHECKS=true > R=[email protected], [email protected] > > Change-Id: Id0093c2bc69e5bc1d3b7147b7bbcc633ed625a45 > Reviewed-on: https://chromium-review.googlesource.com/721919 > Reviewed-by: Adam Klein <[email protected]> > Commit-Queue: Jungshik Shin <[email protected]> > Cr-Commit-Position: refs/branch-heads/6.3@{#7} > Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} > Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432} [email protected],[email protected],[email protected] Change-Id: I8b30f8f9c63b93cb000facbf910ed111ca3f9ab2 No-Presubmit: true No-Tree-Checks: true No-Try: true Bug: chromium:612010, chromium:769706, chromium:771868, v8:2137, v8:6031, v8:6076 Reviewed-on: https://chromium-review.googlesource.com/723324 Reviewed-by: Michael Hablich <[email protected]> Commit-Queue: Michael Hablich <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#18} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
[email protected] Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng Change-Id: I81602ad8b9219df514418f0f8a9368c0a39183f5 Reviewed-on: https://chromium-review.googlesource.com/723700 Reviewed-by: v8 autoroll <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#19} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
…t32Key. Revision: a3de183 BUG=chromium:774459 LOG=N NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: Ia785eaaa6effe18f31ec59fdb9544ffc36b5195f Reviewed-on: https://chromium-review.googlesource.com/723593 Reviewed-by: Camillo Bruni <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#20} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
This reverts commit 3b1cfa3. Reason for revert: ICU was not rolled in 6.3 branch leading invalid-locale test failure (that was added to test an ICU fix). Now, ICU is rolled in 6.3 branch ( https://chromium-review.googlesource.com/c/v8/v8/+/723564 ). Original change's description: > Revert "Merged: Revert "Enable icu-timezone-data by default"" > > This reverts commit d734370. > > Reason for revert: broke some branch builders like https://build.chromium.org/p/client.v8.branches/builders/V8%20arm%20-%20sim%20-%20beta%20branch%20-%20debug > > Original change's description: > > Merged: Revert "Enable icu-timezone-data by default" > > > > This reverts commit d9a2584. > > This is for 6.3 branch (to go with Chromium 63 branch). > > > > Reason for revert: > > > > I'm reverting this CL for a few reasons. #2 is the most significant and > > I should have thought of that before making a switch. Sorry for that. > > > > 1) perf-regression: http://crbug.com/769706 > > 2) http://crbug.com/612010 : ICU timezone update is not propagated to > > zygote process so that new tabs will hold on to an old timezone even > > after a timezone change on Linux and Chrome OS. > > > > 3) http://crbug.com/754053 : OS timezone detection issues on macOS > > 10.13, Ubutu 16, RHEL 7, SuSe Linux 12 or newer. ; it's being fixed. So, > > it actually ok. > > > > 4) http://crbug.com/771868 : timezone wrong in gmail: If it's due to > > > > Revision: 28ef8dc > > > > BUG=chromium:612010,chromium:769706,chromium:771868,v8:2137,v8:6031,v8:6076 > > LOG=N > > NOTRY=true > > NOPRESUBMIT=true > > NOTREECHECKS=true > > R=[email protected], [email protected] > > > > Change-Id: Id0093c2bc69e5bc1d3b7147b7bbcc633ed625a45 > > Reviewed-on: https://chromium-review.googlesource.com/721919 > > Reviewed-by: Adam Klein <[email protected]> > > Commit-Queue: Jungshik Shin <[email protected]> > > Cr-Commit-Position: refs/branch-heads/6.3@{#7} > > Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} > > Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432} > > [email protected],[email protected],[email protected] > > Change-Id: I8b30f8f9c63b93cb000facbf910ed111ca3f9ab2 > No-Presubmit: true > No-Tree-Checks: true > No-Try: true > Bug: chromium:612010, chromium:769706, chromium:771868, v8:2137, v8:6031, v8:6076 > Reviewed-on: https://chromium-review.googlesource.com/723324 > Reviewed-by: Michael Hablich <[email protected]> > Commit-Queue: Michael Hablich <[email protected]> > Cr-Commit-Position: refs/branch-heads/6.3@{#18} > Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} > Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432} [email protected],[email protected],[email protected] Change-Id: I8f712f1e6eb1f14c703ffe9f1d63d23b4b4bb08e No-Presubmit: true No-Tree-Checks: true No-Try: true Bug: chromium:612010, chromium:769706, chromium:771868, v8:2137, v8:6031, v8:6076 Reviewed-on: https://chromium-review.googlesource.com/723962 Reviewed-by: Michael Hablich <[email protected]> Reviewed-by: Adam Klein <[email protected]> Commit-Queue: Jungshik Shin <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#21} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
There are only two changes in the roll. This is to match Chromium M63's ICU in v8's 6.3 branch https://chromium.googlesource.com/chromium/deps/icu/+log/08cb9568..21d33b1a Bug: 770452 Test: See the bug Change-Id: I63007a5ece0222374a02a958bb04a523f66c78c1 No-Presubmit: true No-Tree-Checks: true No-Try: true Reviewed-on: https://chromium-review.googlesource.com/723564 Commit-Queue: Jungshik Shin <[email protected]> Reviewed-by: Adam Klein <[email protected]> Reviewed-by: Michael Hablich <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#22} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
…ag"" This reverts commit 9a132a9. Reason for revert: ICU was not rolled in 6.3 branch leading invalid-locale test failure (that was added to test an ICU fix). Now, ICU is rolled in 6.3 branch ( https://chromium-review.googlesource.com/c/v8/v8/+/723564 ). Original change's description: > Revert "Merged: Correct the misuse of uloc_{to,from}LanguageTag" > > This reverts commit 593b0d8. > > Reason for revert: broke some branch builders like https://build.chromium.org/p/client.v8.branches/builders/V8%20arm%20-%20sim%20-%20beta%20branch%20-%20debug > > Original change's description: > > Merged: Correct the misuse of uloc_{to,from}LanguageTag > > > > Revision: 69bd294 > > > > Merge to 6.3 branch > > > > - remove unused Runtime_GetLanguageTagVariants > > - add test for another related bug (chromium:770452) as well as for > > chromium:770450 . > > > > BUG=chromium:770450,chromium:770452 > > LOG=N > > NOTRY=true > > NOPRESUBMIT=true > > NOTREECHECKS=true > > R=[email protected] > > > > Cq-Include-Trybots: master.tryserver.v8:v8_linux_noi18n_rel_ng > > Change-Id: Ia2c6dd2156c51995fb18228fc3062a86e78d719c > > Reviewed-on: https://chromium-review.googlesource.com/721844 > > Reviewed-by: Adam Klein <[email protected]> > > Commit-Queue: Jungshik Shin <[email protected]> > > Cr-Commit-Position: refs/branch-heads/6.3@{#9} > > Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} > > Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432} > > [email protected],[email protected],[email protected] > > Change-Id: I37018f8241efe1431f453ff55cf8216a5daa66de > No-Presubmit: true > No-Tree-Checks: true > No-Try: true > Bug: chromium:770450, chromium:770452 > Cq-Include-Trybots: master.tryserver.v8:v8_linux_noi18n_rel_ng > Reviewed-on: https://chromium-review.googlesource.com/723323 > Reviewed-by: Michael Hablich <[email protected]> > Commit-Queue: Michael Hablich <[email protected]> > Cr-Commit-Position: refs/branch-heads/6.3@{#17} > Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} > Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432} [email protected],[email protected],[email protected] Change-Id: Ie7eac96859c8053c4f1b41b0a9b4f79a44883295 No-Presubmit: true No-Tree-Checks: true No-Try: true Bug: chromium:770450, chromium:770452 Cq-Include-Trybots: master.tryserver.v8:v8_linux_noi18n_rel_ng Reviewed-on: https://chromium-review.googlesource.com/723608 Reviewed-by: Michael Hablich <[email protected]> Reviewed-by: Adam Klein <[email protected]> Commit-Queue: Jungshik Shin <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#23} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
[email protected] Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng Change-Id: If830953f0ced61f8470d1d1faf1956fffffd39c5 Reviewed-on: https://chromium-review.googlesource.com/724982 Reviewed-by: v8 autoroll <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#24} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
…sing. When skipping inner funcs is enabled, we also track variables for top level funcs. Thus, we also declared the function name for the function scope, even though it was the function scope for a function whose preparsing was aborted. This lead into declaring the function name twice. The fix is to declare the function name only in the success case. The code was "wrong" before too, but this was never a problem, since variable tracking and aborting preparsing were enabled for disjoint sets of functions (aborting preparsing only for top-level, and variable tracking for non-top-level). BUG=v8:5516,chromium:774475 NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true Change-Id: Ie6c321cc834cd946e8843f73916fa7dd75e9cd09 Reviewed-on: https://chromium-review.googlesource.com/720920 Reviewed-by: Toon Verwaest <[email protected]> Commit-Queue: Marja Hölttä <[email protected]> Cr-Original-Commit-Position: refs/heads/master@{#48592}(cherry picked from commit d69159d) Reviewed-on: https://chromium-review.googlesource.com/728000 Reviewed-by: Michael Hablich <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#25} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
Inner functions which called eval, and were the kind of functions that can use `super`, were erroneously not marked as "uses_super_property", leading to downstream crashes when the runtime tried to load the [[HomeObject]] from them. This patch eliminates the public Scope::uses_super_property() API and ensures that callers always call Scope::NeedsHomeObject() instead. This is a minimal fix designed for easy merging; it's likely that in the long run we should remove most mentions of "uses super property" and replace them with "needs home object" for clarity. NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true Bug: v8:5516, chromium:774994 Change-Id: Id269dd33e35bd40f6b59a3d3e19330687afa64f8 Reviewed-on: https://chromium-review.googlesource.com/721879 Reviewed-by: Marja Hölttä <[email protected]> Commit-Queue: Adam Klein <[email protected]> Cr-Original-Commit-Position: refs/heads/master@{#48619}(cherry picked from commit 94a71d7) Reviewed-on: https://chromium-review.googlesource.com/727762 Reviewed-by: Michael Hablich <[email protected]> Commit-Queue: Marja Hölttä <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#26} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
[email protected] Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng Change-Id: I5008b819ac6a5b89af2f8981268d6364bc545b66 Reviewed-on: https://chromium-review.googlesource.com/728040 Reviewed-by: v8 autoroll <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#27} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
This cherry-picks: https://chromium-review.googlesource.com/c/v8/v8/+/725706 https://chromium-review.googlesource.com/c/v8/v8/+/727800 https://chromium-review.googlesource.com/c/v8/v8/+/728023 NOTRY=true NOPRESUBMIT=true Bug: v8:6917 Change-Id: If89a30f9d6548d9402ebad6bcef5b0d5f50802b2 Reviewed-on: https://chromium-review.googlesource.com/727205 Reviewed-by: Sergiy Byelozyorov <[email protected]> Commit-Queue: Michael Achenbach <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#28} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
[email protected] Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng Change-Id: Ife4392a3dd9e6ca19a71fa23927e23f8b52b41f2 Reviewed-on: https://chromium-review.googlesource.com/727943 Reviewed-by: v8 autoroll <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#29} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
Change-Id: Ibf07f73faf67500568c7631179652b0e6d478dcd Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3571206 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/10.0@{v8#21} Cr-Branched-From: 6ea73a7-refs/heads/10.0.139@{#1} Cr-Branched-From: ccc6890-refs/heads/main@{#79131}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: Ief1d6c18322c980bfabd509bf36583ac96841aa1 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3168358 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/9.5@{v8#21} Cr-Branched-From: 4a03d61-refs/heads/9.5.172@{#1} Cr-Branched-From: 9a60704-refs/heads/main@{#76741}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
R=[email protected] (cherry picked from commit 5e3abf4) Bug: chromium:1335523 Change-Id: I194915b37ea27b3d7d1a9f2367b6935b8a3bff39 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3702233 Commit-Queue: Andreas Haas <[email protected]> Reviewed-by: Thibaud Michaud <[email protected]> Cr-Original-Commit-Position: refs/heads/main@{#81095} Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3726213 Cr-Commit-Position: refs/branch-heads/10.2@{v8#21} Cr-Branched-From: 374091f-refs/heads/10.2.154@{#1} Cr-Branched-From: f0c353f-refs/heads/main@{#79976}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
Revision: 49d1520 BUG=skia:10306 NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true Change-Id: Ie6fe5cd4c935fca42cd0c009596e3f39de0a60a3 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3714947 Reviewed-by: Deepti Gandluri <[email protected]> Commit-Queue: Liviu Rau <[email protected]> Cr-Commit-Position: refs/branch-heads/10.4@{v8#21} Cr-Branched-From: b1413ed-refs/heads/10.4.132@{#1} Cr-Branched-From: 9d0a093-refs/heads/main@{#80972}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
This reverts commit a0c79b4. Reason for revert: trigger Skia branch rollers Original change's description: > Merged: White space to trigger Skia branch rollers > > Revision: 49d1520 > > BUG=skia:10306 > NOTRY=true > NOPRESUBMIT=true > NOTREECHECKS=true > > Change-Id: Ie6fe5cd4c935fca42cd0c009596e3f39de0a60a3 > Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3714947 > Reviewed-by: Deepti Gandluri <[email protected]> > Commit-Queue: Liviu Rau <[email protected]> > Cr-Commit-Position: refs/branch-heads/10.4@{v8#21} > Cr-Branched-From: b1413ed-refs/heads/10.4.132@{#1} > Cr-Branched-From: 9d0a093-refs/heads/main@{#80972} Bug: skia:10306 Change-Id: I9410deecbaa25988c384e4b53854ada49456c6a6 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3738262 Commit-Queue: Liviu Rau <[email protected]> Bot-Commit: Rubber Stamper <[email protected]> Cr-Commit-Position: refs/branch-heads/10.4@{v8#27} Cr-Branched-From: b1413ed-refs/heads/10.4.132@{#1} Cr-Branched-From: 9d0a093-refs/heads/main@{#80972}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
Change-Id: Id593446c66d9a78ea05ccdb228a3c083fdd5b0c7 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3447521 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/9.8@{v8#21} Cr-Branched-From: e218afa-refs/heads/9.8.177@{#1} Cr-Branched-From: 86ebfc9-refs/heads/main@{#78312}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
Revision: 574ca6b BUG=chromium:1234770 NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: I15af5a94e89b54c2a540442c3544ed459b832e0a Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3080564 Reviewed-by: Lutz Vahl <[email protected]> Commit-Queue: Georg Neis <[email protected]> Cr-Commit-Position: refs/branch-heads/9.3@{v8#21} Cr-Branched-From: 7744dce-refs/heads/9.3.345@{#1} Cr-Branched-From: 4b6b4ca-refs/heads/master@{#75728}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
Change-Id: I4d3eafa27e6ea7e2aaca51355123be27c34b190c Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3303899 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/9.7@{v8#21} Cr-Branched-From: 49162da-refs/heads/9.7.106@{#1} Cr-Branched-From: a7e9b8f-refs/heads/main@{#77674}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
Revision: 47ea5fb BUG=v8:7790 NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: I760cf08b7bc4d54db96b037729cf6d3691c1d87a Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2505762 Reviewed-by: Ross McIlroy <[email protected]> Cr-Commit-Position: refs/branch-heads/8.7@{v8#21} Cr-Branched-From: 0d81cd7-refs/heads/8.7.220@{#1} Cr-Branched-From: 942c2ef-refs/heads/master@{#70196}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
Change-Id: Ic3448c3eb47e9a0b13974a7ed7370048dd09e783 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4193351 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/11.0@{v8#21} Cr-Branched-From: 06097c6-refs/heads/11.0.226@{#1} Cr-Branched-From: 6bf3344-refs/heads/main@{#84857}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
[email protected] Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng Change-Id: Iaed3e00ac22c82a56c310e0ce8ae7a7a37499ca5 Reviewed-on: https://chromium-review.googlesource.com/586748 Reviewed-by: v8 autoroll <[email protected]> Cr-Commit-Position: refs/branch-heads/6.1@{v8#21} Cr-Branched-From: 1bf2e10-refs/heads/6.1.534@{#1} Cr-Branched-From: e825c43-refs/heads/master@{#46746}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
Merged: Enable typed funcrefs when WasmGC Origin Trial is enabled (commit 5aae428) Merged: Reland "[wasm-gc] Make GC and inlining configurable by Origin Trial" (commit 2833c2e) Bug: v8:7748, chromium:1424421, v8:13810 Change-Id: I867270beec1fb853d8f5aa7db56ed46c66f6c7e9 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4350578 Reviewed-by: Jakob Kummerow <[email protected]> Cr-Commit-Position: refs/branch-heads/11.2@{v8#21} Cr-Branched-From: 755511a-refs/heads/11.2.214@{#1} Cr-Branched-From: e6b1cce-refs/heads/main@{#86014}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
Change-Id: Id720e329fe502a987b7f419e560463a06ebae61c Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4293997 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/11.1@{v8#21} Cr-Branched-From: c77793a-refs/heads/11.1.277@{#1} Cr-Branched-From: 95b79bf-refs/heads/main@{#85479}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
Change-Id: I61ac5b3f2bdf0dce70931ec5767ed0de8038073d Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4193359 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/10.9@{v8#21} Cr-Branched-From: 8ade6bf-refs/heads/10.9.194@{#1} Cr-Branched-From: 9ff2515-refs/heads/main@{#84164}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
…ndian Revision: 34bc3cb NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected], [email protected] Bug: v8:6829 Change-Id: Ib84076d1a5193ca3bb792f19bf3c90a26df7ef70 Reviewed-on: https://chromium-review.googlesource.com/668657 Reviewed-by: Yang Guo <[email protected]> Cr-Commit-Position: refs/branch-heads/6.2@{v8#21} Cr-Branched-From: efa2ac4-refs/heads/6.2.414@{#1} Cr-Branched-From: a861ebb-refs/heads/master@{#47693}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
Change-Id: Ifd716f27c77c9b685bb2f27188dd9c2bc1839546 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4544580 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/11.4@{v8#21} Cr-Branched-From: 8a8a1e7-refs/heads/11.4.183@{#1} Cr-Branched-From: 5483d8e-refs/heads/main@{#87241}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
…rences in Scavenger. Revision: 26bc590 BUG=v8:5807,chromium:704342 LOG=N NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: I9a5c7c20b4cb6ab070639ce88d99730c3421f81c Reviewed-on: https://chromium-review.googlesource.com/486279 Reviewed-by: Michael Lippautz <[email protected]> Cr-Commit-Position: refs/branch-heads/5.9@{v8#21} Cr-Branched-From: fe9bb7e-refs/heads/5.9.211@{#1} Cr-Branched-From: 70ad237-refs/heads/master@{#44591}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
[email protected] Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng Change-Id: I62cae58e0f8f437cb9d7e1eaece34921905f59e5 Reviewed-on: https://chromium-review.googlesource.com/838021 Reviewed-by: v8 autoroll <[email protected]> Cr-Commit-Position: refs/branch-heads/6.4@{v8#21} Cr-Branched-From: 0407506-refs/heads/6.4.388@{#1} Cr-Branched-From: a5fc4e0-refs/heads/master@{#49724}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
Merged: [ia32] Fix invalid DCHECK on cmpw with immediate. Revision: 6560707 Merged: [zone] Fix propagation of OOM in Zone::NewSegment. Revision: fb887b8 Merged: [bootstrapper] Bump stack requirement for native compilation. Revision: 72975d3 BUG=v8:6061,v8:6063,v8:6066 LOG=N NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected], [email protected] Review-Url: https://codereview.chromium.org/2741753004 Cr-Commit-Position: refs/branch-heads/5.8@{v8#21} Cr-Branched-From: eda659c-refs/heads/5.8.283@{#1} Cr-Branched-From: 4310cd0-refs/heads/master@{#43429}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
Change-Id: I4b55c51a57e10e493e20b852c37a747bf897f33e Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4514944 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/11.3@{v8#21} Cr-Branched-From: b0a3a06-refs/heads/11.3.244@{#1} Cr-Branched-From: 0326cf6-refs/heads/main@{#86647}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
Jun 24, 2023
Change-Id: I6c85f519fd2970bd77861e5a6ee7ea0bedeb377c Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4640493 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/11.5@{v8#21} Cr-Branched-From: 0c4044b-refs/heads/11.5.150@{#1} Cr-Branched-From: b71d303-refs/heads/main@{#87781}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
Jul 25, 2023
Change-Id: I198e11351d477b327853861ac5d79f390e2d4567 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4711108 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/11.6@{v8#21} Cr-Branched-From: e29c028-refs/heads/11.6.189@{v8#3} Cr-Branched-From: 95cbef2-refs/heads/main@{#88340}
reposynch Bot
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
Sep 3, 2023
Change-Id: I829b1ee090c4420faa765e8b3b8da94fb6e9f8f4 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4820409 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/11.7@{v8#21} Cr-Branched-From: fe60869-refs/heads/11.7.439@{#1} Cr-Branched-From: aeb4552-refs/heads/main@{#89415}
reposynch Bot
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
Oct 1, 2023
Change-Id: Ib19f8f4755369517b7ae0f9098783ca1c061341e Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4898729 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/11.8@{v8#21} Cr-Branched-From: 935bdbf-refs/heads/11.8.172@{#1} Cr-Branched-From: b82a911-refs/heads/main@{#89779}
reposynch Bot
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
Dec 17, 2023
Change-Id: I98d725f25fe3c407d7235b3a6b90c9eb4a40fe9e Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5115359 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/12.1@{v8#21} Cr-Branched-From: b74ef6f-refs/heads/12.1.285@{#1} Cr-Branched-From: 32857fb-refs/heads/main@{#91313}
reposynch Bot
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
Dec 17, 2023
Change-Id: I504556db58c3c803609ef8483024b568def10da6 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5115700 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/12.0@{v8#21} Cr-Branched-From: ed7b4ca-refs/heads/12.0.267@{#1} Cr-Branched-From: 210e75b-refs/heads/main@{#90651}
reposynch Bot
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
Feb 4, 2024
Change-Id: I66e8d236dea18056b67532b04980ff36cae34676 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5249163 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/12.2@{v8#21} Cr-Branched-From: 6eb5a96-refs/heads/12.2.281@{#1} Cr-Branched-From: 44cf56d-refs/heads/main@{#91934}
hubot
pushed a commit
that referenced
this pull request
Mar 20, 2024
Change-Id: I45ed439b8a983721183f534771ff908bf7d47b98 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5383412 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/12.3@{#21} Cr-Branched-From: a86e197-refs/heads/12.3.219@{#1} Cr-Branched-From: 21869f7-refs/heads/main@{#92385}
hubot
pushed a commit
that referenced
this pull request
Apr 9, 2024
Change-Id: I6c5a39690dc315316fe4ded61c8728d247c4eaa9 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5438473 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/12.4@{#21} Cr-Branched-From: 309640d-refs/heads/12.4.254@{#1} Cr-Branched-From: 5dc2470-refs/heads/main@{#92862}
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
@mpszone