Version 6.2.238 - #17
Closed
NarayanaVishali wants to merge 1 commit into
Closed
NarayanaVishali wants to merge 1 commit into
NarayanaVishali wants to merge 1 commit into
Conversation
Performance and stability improvements on all platforms. [email protected] Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng Change-Id: I93c472d3b286a2ae7e2694880ba2f6f0d77ebc83 Reviewed-on: https://chromium-review.googlesource.com/614983 Reviewed-by: v8 autoroll <[email protected]> Cr-Commit-Position: refs/heads/6.2.238@{#1} Cr-Branched-From: 13a990a-refs/heads/master@{#47351}
hubot
pushed a commit
that referenced
this pull request
Sep 14, 2017
Revision: 1aead19 NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Bug: v8:6826 Change-Id: I1232115751cef914f4f1d01898aea25d4ddeef9c Reviewed-on: https://chromium-review.googlesource.com/666925 Commit-Queue: Myles Borins <[email protected]> Reviewed-by: Michael Hablich <[email protected]> Cr-Commit-Position: refs/branch-heads/6.2@{#17} Cr-Branched-From: efa2ac4-refs/heads/6.2.414@{#1} Cr-Branched-From: a861ebb-refs/heads/master@{#47693}
hubot
pushed a commit
that referenced
this pull request
Oct 17, 2017
This reverts commit 593b0d8. Reason for revert: broke some branch builders like https://build.chromium.org/p/client.v8.branches/builders/V8%20arm%20-%20sim%20-%20beta%20branch%20-%20debug Original change's description: > Merged: Correct the misuse of uloc_{to,from}LanguageTag > > Revision: 69bd294 > > Merge to 6.3 branch > > - remove unused Runtime_GetLanguageTagVariants > - add test for another related bug (chromium:770452) as well as for > chromium:770450 . > > BUG=chromium:770450,chromium:770452 > LOG=N > NOTRY=true > NOPRESUBMIT=true > NOTREECHECKS=true > R=[email protected] > > Cq-Include-Trybots: master.tryserver.v8:v8_linux_noi18n_rel_ng > Change-Id: Ia2c6dd2156c51995fb18228fc3062a86e78d719c > Reviewed-on: https://chromium-review.googlesource.com/721844 > Reviewed-by: Adam Klein <[email protected]> > Commit-Queue: Jungshik Shin <[email protected]> > Cr-Commit-Position: refs/branch-heads/6.3@{#9} > Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} > Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432} [email protected],[email protected],[email protected] Change-Id: I37018f8241efe1431f453ff55cf8216a5daa66de No-Presubmit: true No-Tree-Checks: true No-Try: true Bug: chromium:770450, chromium:770452 Cq-Include-Trybots: master.tryserver.v8:v8_linux_noi18n_rel_ng Reviewed-on: https://chromium-review.googlesource.com/723323 Reviewed-by: Michael Hablich <[email protected]> Commit-Queue: Michael Hablich <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#17} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
hubot
pushed a commit
that referenced
this pull request
Oct 18, 2017
…ag"" This reverts commit 9a132a9. Reason for revert: ICU was not rolled in 6.3 branch leading invalid-locale test failure (that was added to test an ICU fix). Now, ICU is rolled in 6.3 branch ( https://chromium-review.googlesource.com/c/v8/v8/+/723564 ). Original change's description: > Revert "Merged: Correct the misuse of uloc_{to,from}LanguageTag" > > This reverts commit 593b0d8. > > Reason for revert: broke some branch builders like https://build.chromium.org/p/client.v8.branches/builders/V8%20arm%20-%20sim%20-%20beta%20branch%20-%20debug > > Original change's description: > > Merged: Correct the misuse of uloc_{to,from}LanguageTag > > > > Revision: 69bd294 > > > > Merge to 6.3 branch > > > > - remove unused Runtime_GetLanguageTagVariants > > - add test for another related bug (chromium:770452) as well as for > > chromium:770450 . > > > > BUG=chromium:770450,chromium:770452 > > LOG=N > > NOTRY=true > > NOPRESUBMIT=true > > NOTREECHECKS=true > > R=[email protected] > > > > Cq-Include-Trybots: master.tryserver.v8:v8_linux_noi18n_rel_ng > > Change-Id: Ia2c6dd2156c51995fb18228fc3062a86e78d719c > > Reviewed-on: https://chromium-review.googlesource.com/721844 > > Reviewed-by: Adam Klein <[email protected]> > > Commit-Queue: Jungshik Shin <[email protected]> > > Cr-Commit-Position: refs/branch-heads/6.3@{#9} > > Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} > > Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432} > > [email protected],[email protected],[email protected] > > Change-Id: I37018f8241efe1431f453ff55cf8216a5daa66de > No-Presubmit: true > No-Tree-Checks: true > No-Try: true > Bug: chromium:770450, chromium:770452 > Cq-Include-Trybots: master.tryserver.v8:v8_linux_noi18n_rel_ng > Reviewed-on: https://chromium-review.googlesource.com/723323 > Reviewed-by: Michael Hablich <[email protected]> > Commit-Queue: Michael Hablich <[email protected]> > Cr-Commit-Position: refs/branch-heads/6.3@{#17} > Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} > Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432} [email protected],[email protected],[email protected] Change-Id: Ie7eac96859c8053c4f1b41b0a9b4f79a44883295 No-Presubmit: true No-Tree-Checks: true No-Try: true Bug: chromium:770450, chromium:770452 Cq-Include-Trybots: master.tryserver.v8:v8_linux_noi18n_rel_ng Reviewed-on: https://chromium-review.googlesource.com/723608 Reviewed-by: Michael Hablich <[email protected]> Reviewed-by: Adam Klein <[email protected]> Commit-Queue: Jungshik Shin <[email protected]> Cr-Commit-Position: refs/branch-heads/6.3@{#23} Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1} Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
hubot
pushed a commit
that referenced
this pull request
Dec 15, 2017
[email protected] Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng Change-Id: I49e96da5f16f82845022db58019bd163a65c7a5d Reviewed-on: https://chromium-review.googlesource.com/829233 Reviewed-by: v8 autoroll <[email protected]> Cr-Commit-Position: refs/branch-heads/6.4@{#17} Cr-Branched-From: 0407506-refs/heads/6.4.388@{#1} Cr-Branched-From: a5fc4e0-refs/heads/master@{#49724}
hubot
pushed a commit
that referenced
this pull request
Jan 31, 2018
Even though most compilers accept the method to be private, gcc 4.8 complains about this (which seems to be correct). Thus we make this method protected. R=[email protected] Change-Id: Ia49b2ddebe1ced7529d4943107a76a909c355b73 Reviewed-on: https://chromium-review.googlesource.com/890449 Reviewed-by: Franziska Hinkelmann <[email protected]> Commit-Queue: Clemens Hammacher <[email protected]> Cr-Commit-Position: refs/heads/master@{#50931}(cherry picked from commit 74915b7) NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true Change-Id: I3c7100bf65785ad2cc01ba5da316a1d90e56d124 Reviewed-on: https://chromium-review.googlesource.com/895305 Reviewed-by: Clemens Hammacher <[email protected]> Commit-Queue: Clemens Hammacher <[email protected]> Cr-Commit-Position: refs/branch-heads/6.5@{#17} Cr-Branched-From: 73c55f5-refs/heads/6.5.254@{#1} Cr-Branched-From: 594a1a0-refs/heads/master@{#50664}
hubot
pushed a commit
that referenced
this pull request
Mar 12, 2018
Some compilers complain that ConvertReturn is private in SimulatorBase, but used in the arm64 Simulator. This CL fixes this by making ConvertReturn protected. R=[email protected] NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true Bug: v8:7541 Change-Id: I0bf5319c6363f2d8c6426d198c5cffc0d1f6619e Originally-reviewed-on: https://chromium-review.googlesource.com/955689 Reviewed-on: https://chromium-review.googlesource.com/958164 Reviewed-by: Andreas Haas <[email protected]> Commit-Queue: Clemens Hammacher <[email protected]> Cr-Commit-Position: refs/branch-heads/6.6@{#17} Cr-Branched-From: d500271-refs/heads/6.6.346@{#1} Cr-Branched-From: 265ef0b-refs/heads/master@{#51624}
hubot
pushed a commit
that referenced
this pull request
Apr 17, 2018
[email protected] Cq-Include-Trybots: luci.chromium.try:linux_chromium_rel_ng Change-Id: I98cadb02d2f4f961b8c424c82099cae9cda13da6 Reviewed-on: https://chromium-review.googlesource.com/1013310 Reviewed-by: v8 autoroll <[email protected]> Cr-Commit-Position: refs/branch-heads/6.7@{#17} Cr-Branched-From: 8457e81-refs/heads/6.7.288@{#2} Cr-Branched-From: e921be5-refs/heads/master@{#52547}
dumganhar
pushed a commit
to dumganhar/v8_fork
that referenced
this pull request
Jun 5, 2018
…chset v8#17 id:320001 of https://codereview.chromium.org/2724373002/ ) Reason for revert: Breaks noi18n: https://build.chromium.org/p/client.v8/builders/V8%20Linux%20-%20noi18n%20-%20debug/builds/13314 Original issue's description: > [date] Add ICU backend for timezone info behind a flag > > This patch implements a timezone backend which is based on ICU, rather > than operating system calls. It can be turned on by passing the > --icu-timezone-data flag. The goal here is to take advantage of ICU's > data, which is more complete than the data that some system calls expose. > For example, without any special code, this patch fixes the time zone > of Lord Howe Island to have a correct 30 minute DST offset, rather than > 60 minutes as the OS backends assume it to have. > > Unfortunately, the parenthized timezone name in Date.prototype.toString() > differs across platforms. This patch chooses the long timezone name, > which matches Windows behavior and might be the most intelligible, but > the web compatibility impact is unclear. > > BUG=v8:6031,v8:2137,v8:6076 > > Review-Url: https://codereview.chromium.org/2724373002 > Cr-Commit-Position: refs/heads/master@{#44562} > Committed: https://chromium.googlesource.com/v8/v8/+/b213f2399038a615cdfbfa0201cddc113d304018 [email protected],[email protected],[email protected],[email protected] # Skipping CQ checks because original CL landed less than 1 days ago. NOPRESUBMIT=true NOTREECHECKS=true NOTRY=true BUG=v8:6031,v8:2137,v8:6076 Review-Url: https://codereview.chromium.org/2811103002 Cr-Commit-Position: refs/heads/master@{#44565}
dumganhar
pushed a commit
to dumganhar/v8_fork
that referenced
this pull request
Jun 5, 2018
…chset #1 id:1 of https://codereview.chromium.org/2811103002/ ) Reason for revert: Reland with tests marked as off in no-i18n mode Original issue's description: > Revert of [date] Add ICU backend for timezone info behind a flag (patchset v8#17 id:320001 of https://codereview.chromium.org/2724373002/ ) > > Reason for revert: > Breaks noi18n: > https://build.chromium.org/p/client.v8/builders/V8%20Linux%20-%20noi18n%20-%20debug/builds/13314 > > Original issue's description: > > [date] Add ICU backend for timezone info behind a flag > > > > This patch implements a timezone backend which is based on ICU, rather > > than operating system calls. It can be turned on by passing the > > --icu-timezone-data flag. The goal here is to take advantage of ICU's > > data, which is more complete than the data that some system calls expose. > > For example, without any special code, this patch fixes the time zone > > of Lord Howe Island to have a correct 30 minute DST offset, rather than > > 60 minutes as the OS backends assume it to have. > > > > Unfortunately, the parenthized timezone name in Date.prototype.toString() > > differs across platforms. This patch chooses the long timezone name, > > which matches Windows behavior and might be the most intelligible, but > > the web compatibility impact is unclear. > > > > BUG=v8:6031,v8:2137,v8:6076 > > > > Review-Url: https://codereview.chromium.org/2724373002 > > Cr-Commit-Position: refs/heads/master@{#44562} > > Committed: https://chromium.googlesource.com/v8/v8/+/b213f2399038a615cdfbfa0201cddc113d304018 > > [email protected],[email protected],[email protected],[email protected] > # Skipping CQ checks because original CL landed less than 1 days ago. > NOPRESUBMIT=true > NOTREECHECKS=true > NOTRY=true > BUG=v8:6031,v8:2137,v8:6076 > > Review-Url: https://codereview.chromium.org/2811103002 > Cr-Commit-Position: refs/heads/master@{#44565} > Committed: https://chromium.googlesource.com/v8/v8/+/13ad50811024ace5623d5d4d13cea4ef21f4affd [email protected],[email protected],[email protected],[email protected] BUG=v8:6031,v8:2137,v8:6076 Review-Url: https://codereview.chromium.org/2813863002 Cr-Commit-Position: refs/heads/master@{#44575}
dumganhar
pushed a commit
to dumganhar/v8_fork
that referenced
this pull request
Jun 5, 2018
Revision: 6803eef BUG=v8:6435 LOG=N NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected], [email protected] Change-Id: Ib4a9c711413f616163fa0d508c063dc5fb561aa7 Reviewed-on: https://chromium-review.googlesource.com/522987 Reviewed-by: Adam Klein <[email protected]> Cr-Commit-Position: refs/branch-heads/6.0@{v8#17} Cr-Branched-From: 97dbf62-refs/heads/6.0.286@{#1} Cr-Branched-From: 12e6f1c-refs/heads/master@{#45439}
hubot
pushed a commit
that referenced
this pull request
Jun 7, 2018
Revision: 4e6bcd2 Bug=v8:7821,v8:7803 NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true Change-Id: I88186b45a98385d8bc65189ad5778523806e877b Reviewed-on: https://chromium-review.googlesource.com/1079572 Reviewed-by: Clemens Hammacher <[email protected]> Commit-Queue: Clemens Hammacher <[email protected]> Cr-Commit-Position: refs/branch-heads/6.8@{#17} Cr-Branched-From: 44d7d7d-refs/heads/6.8.275@{#1} Cr-Branched-From: 5754f66-refs/heads/master@{#53286}
kisg
pushed a commit
to paul99/v8mips
that referenced
this pull request
Jul 4, 2018
This reverts commit 52a10e5. Reason for revert: https://test-results.appspot.com/data/layout_results/V8-Blink_Linux_64__dbg_/12434/layout-test-results/results.html Crash e.g. in http/tests/devtools/oopif/oopif-performance-cpu-profiles.js crash log for devtools (pid <unknown>): STDOUT: <empty> STDERR: STDERR: STDERR: # STDERR: # Fatal error in ../../v8/src/compilation-dependencies.cc, line 281 STDERR: # Debug check failed: descriptor == owner->LastAdded() (10 vs. 22). STDERR: # STDERR: # STDERR: # STDERR: #FailureMessage Object: 0x7fff86878630#0 0x0000031c642c base::debug::StackTrace::StackTrace() STDERR: #1 0x0000046a56bb gin::(anonymous namespace)::PrintStackTrace() STDERR: v8#2 0x00000469c528 V8_Fatal() STDERR: v8#3 0x00000469c285 v8::base::(anonymous namespace)::DefaultDcheckHandler() STDERR: v8#4 0x000001cc5253 v8::internal::CompilationDependencies::DependOnFieldType() STDERR: v8#5 0x000001cdcc46 v8::internal::compiler::AccessInfoFactory::ComputePropertyAccessInfo() STDERR: v8#6 0x000001cde661 v8::internal::compiler::AccessInfoFactory::ComputePropertyAccessInfos() STDERR: v8#7 0x000001dd982b v8::internal::compiler::JSNativeContextSpecialization::ReduceNamedAccess() STDERR: v8#8 0x000001ddb715 v8::internal::compiler::JSNativeContextSpecialization::ReduceNamedAccessFromNexus() STDERR: v8#9 0x000001dd656d v8::internal::compiler::JSNativeContextSpecialization::ReduceJSLoadNamed() STDERR: v8#10 0x000001d53872 v8::internal::compiler::GraphReducer::Reduce() STDERR: v8#11 0x000001d534a5 v8::internal::compiler::GraphReducer::ReduceTop() STDERR: v8#12 0x000001d52e58 v8::internal::compiler::GraphReducer::ReduceNode() STDERR: v8#13 0x000001e4c201 v8::internal::compiler::InliningPhase::Run() STDERR: v8#14 0x000001e44f79 v8::internal::compiler::PipelineImpl::Run<>() STDERR: v8#15 0x000001e41058 v8::internal::compiler::PipelineImpl::CreateGraph() STDERR: v8#16 0x000001e40c75 v8::internal::compiler::PipelineCompilationJob::PrepareJobImpl() STDERR: v8#17 0x000001ccd437 v8::internal::OptimizedCompilationJob::PrepareJob() STDERR: v8#18 0x000001cd071e v8::internal::(anonymous namespace)::GetOptimizedCode() STDERR: v8#19 0x000001cd0c6f v8::internal::Compiler::CompileOptimized() STDERR: v8#20 0x00000231fb62 v8::internal::__RT_impl_Runtime_CompileOptimized_Concurrent() STDERR: v8#21 0x00000288e535 <unknown> Original change's description: > [turbofan] Rewrite CompilationDependencies > > Instead of installing code dependencies during graph reduction, > install them after code generation. > > Bug: v8:7902, v8:7790 > Change-Id: I8a3798254abb5b9ec7c295a1592aeb6b51f24c7a > Reviewed-on: https://chromium-review.googlesource.com/1119913 > Commit-Queue: Georg Neis <[email protected]> > Reviewed-by: Jaroslav Sevcik <[email protected]> > Reviewed-by: Michael Starzinger <[email protected]> > Cr-Commit-Position: refs/heads/master@{#54170} [email protected],[email protected],[email protected] Change-Id: Ic58c2bfadbd34bb6ba7dc0d2b74871cc90b0a74f No-Presubmit: true No-Tree-Checks: true No-Try: true Bug: v8:7902, v8:7790 Reviewed-on: https://chromium-review.googlesource.com/1125680 Reviewed-by: Yang Guo <[email protected]> Commit-Queue: Yang Guo <[email protected]> Cr-Commit-Position: refs/heads/master@{#54192}
hubot
pushed a commit
that referenced
this pull request
Jul 31, 2018
…390.cc Revision: 2a62485 NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: I34864b2345051e5d5f696e547127af16139adcc2 Reviewed-on: https://chromium-review.googlesource.com/1155650 Reviewed-by: John Barboza <[email protected]> Commit-Queue: Junliang Yan <[email protected]> Cr-Commit-Position: refs/branch-heads/6.9@{#17} Cr-Branched-From: d7b61ab-refs/heads/6.9.427@{#1} Cr-Branched-From: b7e108d-refs/heads/master@{#54504}
peterwmwong
pushed a commit
to peterwmwong/v8
that referenced
this pull request
Dec 9, 2018
This reverts commit caed2cc. Reason for revert: Breaks layout tests, e.g. https://test-results.appspot.com/data/layout_results/V8-Blink_Linux_64__dbg_/14924/webkit_layout_tests%20%28with%20patch%29/layout-test-results/results.html crash log for renderer (pid <unknown>): STDOUT: <empty> STDERR: STDERR: STDERR: # STDERR: # Fatal error in ../../v8/src/base/platform/elapsed-timer.h, line 24 STDERR: # Debug check failed: !IsStarted(). STDERR: # STDERR: # STDERR: # STDERR: #FailureMessage Object: 0x7ffc46707640#0 0x565409263b6f base::debug::StackTrace::StackTrace() STDERR: #1 0x56540a8a32fb gin::(anonymous namespace)::PrintStackTrace() STDERR: #2 0x56540a8980d8 V8_Fatal() STDERR: #3 0x56540a897e35 v8::base::(anonymous namespace)::DefaultDcheckHandler() STDERR: v8#4 0x565407971f02 v8::base::ElapsedTimer::Start() STDERR: v8#5 0x565407d08edf v8::internal::TimedHistogram::Start() STDERR: v8#6 0x565407e500d5 v8::internal::IncrementalMarking::AdvanceIncrementalMarkingOnAllocation() STDERR: v8#7 0x565407e4f977 v8::internal::IncrementalMarking::Observer::Step() STDERR: v8#8 0x565407e48092 v8::internal::AllocationObserver::AllocationStep() STDERR: v8#9 0x565407eb0751 v8::internal::SpaceWithLinearArea::InlineAllocationStep() STDERR: v8#10 0x565407eb3e44 v8::internal::NewSpace::EnsureAllocation() STDERR: v8#11 0x565407e258ff v8::internal::NewSpace::AllocateRaw() STDERR: v8#12 0x565407e06b2d v8::internal::Heap::AllocateRaw() STDERR: v8#13 0x565407e432ef v8::internal::Heap::AllocateRawWithLightRetry() STDERR: v8#14 0x565407e433cf v8::internal::Heap::AllocateRawWithRetryOrFail() STDERR: v8#15 0x565407e04d48 v8::internal::Factory::NewFixedArrayWithFiller() STDERR: v8#16 0x565407fd6339 v8::internal::HashTable<>::New() STDERR: v8#17 0x565407fd7be8 v8::internal::HashTable<>::EnsureCapacity() STDERR: v8#18 0x565407fc7e95 v8::internal::Dictionary<>::Add() STDERR: v8#19 0x565407fcf453 v8::internal::BaseNameDictionary<>::Add() STDERR: v8#20 0x565407f89ee4 v8::internal::LookupIterator::ApplyTransitionToDataProperty() STDERR: v8#21 0x5654080036e2 v8::internal::Object::AddDataProperty() STDERR: v8#22 0x56540793061f v8::internal::(anonymous namespace)::DefineDataProperty() STDERR: v8#23 0x56540792da59 v8::internal::(anonymous namespace)::InstantiateObject() STDERR: v8#24 0x56540792b75a v8::internal::(anonymous namespace)::InstantiateFunction() STDERR: v8#25 0x56540792b4db v8::internal::ApiNatives::InstantiateFunction() STDERR: v8#26 0x5654079594bf v8::FunctionTemplate::GetFunction() STDERR: v8#27 0x56540a7af74e blink::V8ObjectConstructor::CreateInterfaceObject() STDERR: v8#28 0x56540a7afe01 blink::V8PerContextData::ConstructorForTypeSlowCase() STDERR: v8#29 0x56540a7afdd6 blink::V8PerContextData::ConstructorForTypeSlowCase() STDERR: v8#30 0x56540a7afdd6 blink::V8PerContextData::ConstructorForTypeSlowCase() STDERR: v8#31 0x56540a7afcb4 blink::V8PerContextData::CreateWrapperFromCacheSlowCase() STDERR: v8#32 0x56540a7aef73 blink::V8DOMWrapper::CreateWrapper() STDERR: v8#33 0x56540a7abf6b blink::ScriptWrappable::Wrap() STDERR: v8#34 0x56540a677199 blink::V8Document::documentElementAttributeGetterCallback() STDERR: v8#35 0x565407a0aec3 v8::internal::FunctionCallbackArguments::Call() STDERR: v8#36 0x565407a097be v8::internal::(anonymous namespace)::HandleApiCallHelper<>() STDERR: v8#37 0x565407a0877b v8::internal::Builtins::InvokeApiFunction() STDERR: v8#38 0x565407fe785a v8::internal::Object::GetPropertyWithAccessor() STDERR: v8#39 0x565407fe697e v8::internal::Object::GetProperty() STDERR: v8#40 0x565407ec8c71 v8::internal::LoadIC::Load() STDERR: v8#41 0x565407ed6401 v8::internal::__RT_impl_Runtime_LoadIC_Miss() STDERR: v8#42 0x5654087593f2 <unknown> STDERR: [16162:16185:1122/143518.356897:WARNING:crash_handler_host_linux.cc(341)] Could not translate tid, attempt = 1 retry ... Original change's description: > [heap] Improve embedder tracing during incremental marking > > Add a path into embedder tracing on allocation. This is safe as as Blink > is not allowed to call into V8 during object construction. > > Bug: chromium:843903 > Change-Id: I5af053c3169f5a33778ebce5d7c5c43e4efb1aa4 > Reviewed-on: https://chromium-review.googlesource.com/c/1348749 > Commit-Queue: Michael Lippautz <[email protected]> > Reviewed-by: Ulan Degenbaev <[email protected]> > Cr-Commit-Position: refs/heads/master@{#57757} [email protected],[email protected] Change-Id: Ide2c0b284b52bee17573adcc89f14be4e40dab91 No-Presubmit: true No-Tree-Checks: true No-Try: true Bug: chromium:843903 Reviewed-on: https://chromium-review.googlesource.com/c/1349189 Reviewed-by: Yang Guo <[email protected]> Commit-Queue: Yang Guo <[email protected]> Cr-Commit-Position: refs/heads/master@{#57759}
junhuaw
pushed a commit
to junhuaw/v8
that referenced
this pull request
Feb 6, 2019
Revision: 157af78 BUG=v8:8525 LOG=N NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: I56787a43596dce96f78d33de9aba0a369428fd17 Reviewed-on: https://chromium-review.googlesource.com/c/1370012 Reviewed-by: Adam Klein <[email protected]> Cr-Commit-Position: refs/branch-heads/7.2@{v8#17} Cr-Branched-From: 6acd03c-refs/heads/7.2.502@{#1} Cr-Branched-From: b03041d-refs/heads/master@{#57910}
Teemperor
pushed a commit
to Teemperor/v8
that referenced
this pull request
May 25, 2019
Merged: [proxy] fix has trap check for indices Revision: 11d8358 Merged: [csa] Make JSProxy's CheckGetSetTrapResult bailout for certain names Revision: b9962a9 Merged: [csa] Remove accidental debug print Revision: 58218e5 BUG=chromium:937618 LOG=N NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: I599e9189e66db6ff59cecf1aeb7e1bb99ee059aa Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1523546 Reviewed-by: Georg Neis <[email protected]> Cr-Commit-Position: refs/branch-heads/7.4@{v8#17} Cr-Branched-From: 3e8a733-refs/heads/7.4.288@{#1} Cr-Branched-From: d077f9b-refs/heads/master@{#60039}
billti
pushed a commit
to billti/v8
that referenced
this pull request
Apr 6, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: I42d9a7c28e3230fb79e238eb539b993c45091c78 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2057033 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/8.1@{v8#17} Cr-Branched-From: a4dcd39-refs/heads/8.1.307@{#1} Cr-Branched-From: f22c213-refs/heads/master@{#66031}
Kwizatz
pushed a commit
to AeonGames/v8
that referenced
this pull request
Sep 2, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: I7508fdf18f470e5cdd92ae7d1e60818ead459fbb Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2292976 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/8.5@{v8#17} Cr-Branched-From: a7f8bc4-refs/heads/8.5.210@{#1} Cr-Branched-From: dd58472-refs/heads/master@{#68510}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
Revision: ecafe04 BUG=chromium:997056 LOG=N NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: I45d662124c9f29c0f7494ef8424cefdf14ba431f Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1820938 Reviewed-by: Igor Sheludko <[email protected]> Cr-Commit-Position: refs/branch-heads/7.8@{v8#17} Cr-Branched-From: 73694fd-refs/heads/7.8.279@{#1} Cr-Branched-From: 2314928-refs/heads/master@{#63555}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
…ring This fixes an invalid assumption when emitting code for matching '^' (start of line) in multiline regexps and '\b', '\B' in general. What we used to do: if the current trace's cp_offset (the offset from the current position) was non-zero, we assumed that we were looking at subject string index 1 or greater (i.e.: not at the start of the string or before). This is no longer valid since cp_offsets can now be negative. This CL changes the logic to omit start- and bounds-checks only for strictly positive cp_offsets, where the above assumption still holds. (cherry picked from commit 1990b1e) Tbr: [email protected] No-Try: true No-Presubmit: true No-Treechecks: true Bug: chromium:996391 Change-Id: I79be4fc295c6f0b63e41c13d1e91fdd00f2f2b42 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1771794 Commit-Queue: Erik Corry <[email protected]> Auto-Submit: Jakob Gruber <[email protected]> Reviewed-by: Erik Corry <[email protected]> Cr-Original-Commit-Position: refs/heads/master@{#63424} Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1781680 Reviewed-by: Jakob Gruber <[email protected]> Commit-Queue: Jakob Gruber <[email protected]> Cr-Commit-Position: refs/branch-heads/7.7@{v8#17} Cr-Branched-From: 4035531-refs/heads/7.7.299@{#1} Cr-Branched-From: 1320c91-refs/heads/master@{#62881}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
Revert "Reland "[ptr-compr][arm64] Temporarily enable pointer compression on arm64"" (https://chromium-review.googlesource.com/c/v8/v8/+/1746470) [email protected] Bug: v8:9885 Change-Id: I54b9ad0e0f7050b39e04ade1450adb2ea7eb43bf Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1873686 Reviewed-by: Dan Elphick <[email protected]> Cr-Commit-Position: refs/branch-heads/7.9@{v8#17} Cr-Branched-From: be181e2-refs/heads/7.9.317@{#1} Cr-Branched-From: 0d7889d-refs/heads/master@{#64307}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: I54946f69031892c8567aaa20d2356f3076784342 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1984853 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/8.0@{v8#17} Cr-Branched-From: 69827db-refs/heads/8.0.426@{v8#2} Cr-Branched-From: 2fe1552-refs/heads/master@{#65318}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: I2796c032ae717eb7815c4c8b149cbb4a725a6dd7 Reviewed-on: https://chromium-review.googlesource.com/c/1459276 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/7.3@{v8#17} Cr-Branched-From: 9df9418-refs/heads/7.3.492@{v8#2} Cr-Branched-From: be213cf-refs/heads/master@{#59024}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Cq-Include-Trybots: luci.chromium.try:linux_chromium_rel_ng Change-Id: I5a152b0425daf1f8140e21c36bd539b6991e3120 Reviewed-on: https://chromium-review.googlesource.com/c/1298914 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/7.1@{v8#17} Cr-Branched-From: f70aaa8-refs/heads/7.1.302@{#1} Cr-Branched-From: 1dbcc78-refs/heads/master@{#56462}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: I85ffb7934df927ac20762a4547891aa3d6460961 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1670807 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/7.6@{v8#17} Cr-Branched-From: 2cb2573-refs/heads/7.6.303@{#1} Cr-Branched-From: 201c509-refs/heads/master@{#61902}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: I6c4f7038e0ed641bcc7e9fb13a603d80319d07e8 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1592718 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/7.5@{v8#17} Cr-Branched-From: 35b9bf5-refs/heads/7.5.288@{#1} Cr-Branched-From: 912b391-refs/heads/master@{#60911}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: Idce287afbb86b6e8085d24915ab3b81a186fe613 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2182154 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/8.3@{v8#17} Cr-Branched-From: 1668abd-refs/heads/8.3.110@{#1} Cr-Branched-From: 04a7a68-refs/heads/master@{#66926}
salewski
pushed a commit
to salewski/v8
that referenced
this pull request
Aug 26, 2021
Revision: 346f1b6 BUG=chromium:1238515,v8:11946 NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: I4ad0bcde7cae888f16dbcd88fb38ab0ad62a336b Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3103319 Reviewed-by: Victor Gomes <[email protected]> Commit-Queue: Camillo Bruni <[email protected]> Cr-Commit-Position: refs/branch-heads/9.4@{v8#17} Cr-Branched-From: 3b51863-refs/heads/9.4.146@{#1} Cr-Branched-From: 2890419-refs/heads/master@{#76233}
BenPope
pushed a commit
to BenPope/v8
that referenced
this pull request
Dec 10, 2021
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: I6c49ce474b7de83a9507f362a0b2026715663cf6 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3246094 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/9.6@{v8#17} Cr-Branched-From: 0b7bda0-refs/heads/9.6.180@{#1} Cr-Branched-From: 41a5a24-refs/heads/main@{#77244}
Kwizatz
pushed a commit
to AeonGames/v8
that referenced
this pull request
Jun 27, 2022
Change-Id: I02f9506e91938d5261421bd453cd411920ef91c9 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3572948 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/10.1@{v8#17} Cr-Branched-From: b003970-refs/heads/10.1.124@{#1} Cr-Branched-From: e62f556-refs/heads/main@{#79503}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
This fixes a bug introduced in crrev.com/c/2717308. For JSArray holders, we must observe JSArray::length for bounds checks (in addition to elements.length). JSArray::length cannot reliably be read from the background thread; thus we do a best-effort read there, and verify the result during finalization through a new ArrayIndexIsInBoundsDependency. (cherry picked from commit 1ff04cc) No-Try: true No-Presubmit: true No-Treechecks: true Bug: v8:7790,chromium:1209444 Change-Id: I189df9f58043411ada62f32fe741d4729874d357 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2928509 Commit-Queue: Jakob Gruber <[email protected]> Reviewed-by: Georg Neis <[email protected]> Cr-Original-Commit-Position: refs/heads/master@{#74904} Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2940891 Reviewed-by: Jakob Gruber <[email protected]> Bot-Commit: Rubber Stamper <[email protected]> Cr-Commit-Position: refs/branch-heads/9.2@{v8#17} Cr-Branched-From: 5123834-refs/heads/9.2.230@{#1} Cr-Branched-From: 587a04f-refs/heads/master@{#74656}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: Iff13b785d596949dc0d85ce9235aac38f1f076ea Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2642576 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/8.9@{v8#17} Cr-Branched-From: 16b9bbb-refs/heads/8.9.255@{#1} Cr-Branched-From: d16a2a6-refs/heads/master@{#72039}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
Change-Id: Ic9b3150b98c9ac9328cfeadffab8b86744b61575 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3679982 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/10.3@{v8#17} Cr-Branched-From: 1a8f4ca-refs/heads/10.3.174@{#1} Cr-Branched-From: 8fbefa4-refs/heads/main@{#80471}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
Change-Id: I290ebaeb0941e4ec2f1bfe6e50e3ad8b7a736ff8 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3565175 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/10.0@{v8#17} Cr-Branched-From: 6ea73a7-refs/heads/10.0.139@{#1} Cr-Branched-From: ccc6890-refs/heads/main@{#79131}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: Icf37a893ea211875d9dd3ba4804a530a22983004 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3163622 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/9.5@{v8#17} Cr-Branched-From: 4a03d61-refs/heads/9.5.172@{#1} Cr-Branched-From: 9a60704-refs/heads/main@{#76741}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
Change-Id: Ic4d64cae66e60797af616b23815f067e6d3663f5 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3507196 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/9.9@{v8#17} Cr-Branched-From: c4208cc-refs/heads/9.9.115@{#1} Cr-Branched-From: 19c6fd7-refs/heads/main@{#78682}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
Change-Id: I173402433e55627156a2425c2cb1c3d0c1b40389 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3695972 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/10.2@{v8#17} Cr-Branched-From: 374091f-refs/heads/10.2.154@{#1} Cr-Branched-From: f0c353f-refs/heads/main@{#79976}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
…t promise" The original CL used Object::Set to create the result object of WebAssembly instantiation. However, Object::Set is potentially observable from JavaScript, and therefore required a MicrotasksScope. This CL replaces the use of Object::Set with Object::CreateDataProperty. Original message: This CL switches resolving and rejecting the wasm result promise from the V8-internal API to the external API added in https://chromium-review.googlesource.com/c/v8/v8/+/3695584. This CL can land once Chrome provided an implementation of the callback. R=[email protected] (cherry picked from commit 438dc0d) Bug: v8:12953 Change-Id: If1f252736fd3a13024d4b38adebf468530c59c03 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3714234 Reviewed-by: Jakob Kummerow <[email protected]> Commit-Queue: Andreas Haas <[email protected]> Cr-Original-Commit-Position: refs/heads/main@{#81253} Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3716484 Cr-Commit-Position: refs/branch-heads/10.4@{v8#17} Cr-Branched-From: b1413ed-refs/heads/10.4.132@{#1} Cr-Branched-From: 9d0a093-refs/heads/main@{#80972}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
Change-Id: I7709b56d2ec5b7348679a1ac905c864fda299944 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3411772 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/9.8@{v8#17} Cr-Branched-From: e218afa-refs/heads/9.8.177@{#1} Cr-Branched-From: 86ebfc9-refs/heads/main@{#78312}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: I2da96eb150953d8afac6334f149ead1ba0e06223 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3065576 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/9.3@{v8#17} Cr-Branched-From: 7744dce-refs/heads/9.3.345@{#1} Cr-Branched-From: 4b6b4ca-refs/heads/master@{#75728}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
[email protected] No-Try: true No-Presubmit: true No-Tree-Checks: true Change-Id: I5b6cc839d1ce97e21d0f0b3e96ef69ff0f75a835 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3295746 Bot-Commit: Rubber Stamper <[email protected]> Reviewed-by: Zhi An Ng <[email protected]> Commit-Queue: Liviu Rau <[email protected]> Cr-Commit-Position: refs/branch-heads/9.7@{v8#17} Cr-Branched-From: 49162da-refs/heads/9.7.106@{#1} Cr-Branched-From: a7e9b8f-refs/heads/main@{#77674}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: Ia015f8749cb6d5b22969cf081b297e218a3d6ec3 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2492501 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/8.7@{v8#17} Cr-Branched-From: 0d81cd7-refs/heads/8.7.220@{#1} Cr-Branched-From: 942c2ef-refs/heads/master@{#70196}
unicornx
pushed a commit
to unicornx/v8
that referenced
this pull request
Feb 16, 2023
Change-Id: I8b0056e4d836d3cd70d8fc7350250d3e27bb0d01 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4117882 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/10.9@{v8#17} Cr-Branched-From: 8ade6bf-refs/heads/10.9.194@{aosp-riscv#1} Cr-Branched-From: 9ff2515-refs/heads/main@{#84164}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
Change-Id: I285eea788ed7530286da1c470c40eb6c1a51807c Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4148127 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/11.0@{v8#17} Cr-Branched-From: 06097c6-refs/heads/11.0.226@{#1} Cr-Branched-From: 6bf3344-refs/heads/main@{#84857}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
[email protected] Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng Change-Id: Icb66f2486976fdf8ccdc9231c004933d400ce8f9 Reviewed-on: https://chromium-review.googlesource.com/586109 Reviewed-by: v8 autoroll <[email protected]> Cr-Commit-Position: refs/branch-heads/6.1@{v8#17} Cr-Branched-From: 1bf2e10-refs/heads/6.1.534@{#1} Cr-Branched-From: e825c43-refs/heads/master@{#46746}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
Revision: 3e4952c NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true Change-Id: I2cce6ba6f8fa23a3ff5aec5ff242e1472a78ef6d Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4335219 Reviewed-by: Junliang Yan <[email protected]> Reviewed-by: Toon Verwaest <[email protected]> Commit-Queue: Toon Verwaest <[email protected]> Cr-Commit-Position: refs/branch-heads/11.2@{v8#17} Cr-Branched-From: 755511a-refs/heads/11.2.214@{#1} Cr-Branched-From: e6b1cce-refs/heads/main@{#86014}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
Looking at the representation axis of the types in any of the early compilation phases is not correct, actually looking there at all is not correct (at least these days). NOTRY=true NOPRESUBMIT=true BUG=chromium:642056 [email protected] Review-Url: https://codereview.chromium.org/2307753002 Cr-Commit-Position: refs/branch-heads/5.4@{v8#17} Cr-Branched-From: 5ce2827-refs/heads/5.4.500@{v8#2} Cr-Branched-From: ad07b49-refs/heads/master@{#38841}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
Revision: ca3f487 BUG=chromium:665659 LOG=N NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected], [email protected] Review URL: https://codereview.chromium.org/2527463003 . Cr-Commit-Position: refs/branch-heads/5.6@{v8#17} Cr-Branched-From: bdd3886-refs/heads/5.6.326@{#1} Cr-Branched-From: 879f659-refs/heads/master@{#41014}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
Change-Id: I9d3ef04625a1096539646364494d9ac67407cbc0 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4275725 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/11.1@{v8#17} Cr-Branched-From: c77793a-refs/heads/11.1.277@{#1} Cr-Branched-From: 95b79bf-refs/heads/main@{#85479}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
Change-Id: Id4f14e6a448ca0f696277a4de13f12466eeba469 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4542636 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/11.4@{v8#17} Cr-Branched-From: 8a8a1e7-refs/heads/11.4.183@{#1} Cr-Branched-From: 5483d8e-refs/heads/main@{#87241}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
…ossible. Revision: 385734b BUG=v8:6287 LOG=N NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: I0146566088b94215061f02470a6b8ed24f4f9042 Reviewed-on: https://chromium-review.googlesource.com/484381 Reviewed-by: Leszek Swirski <[email protected]> Cr-Commit-Position: refs/branch-heads/5.9@{v8#17} Cr-Branched-From: fe9bb7e-refs/heads/5.9.211@{#1} Cr-Branched-From: 70ad237-refs/heads/master@{#44591}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
Merged: MIPS[64]: Fix unaligned arguments storage in Wasm-to-interpreter entry Revision: 84ff6e4 Merged: MIPS: Fix int64->int32 lowering in wasm-to-interpeter entry on big-endian archs. Revision: 4f426e1 NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] BUG=v8:6043 Review-Url: https://codereview.chromium.org/2734253003 . Cr-Commit-Position: refs/branch-heads/5.8@{v8#17} Cr-Branched-From: eda659c-refs/heads/5.8.283@{#1} Cr-Branched-From: 4310cd0-refs/heads/master@{#43429}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
Change-Id: If1e83497cf4a998c69c24ab9d77190e293981df9 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4490447 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/11.3@{v8#17} Cr-Branched-From: b0a3a06-refs/heads/11.3.244@{#1} Cr-Branched-From: 0326cf6-refs/heads/main@{#86647}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
Jun 15, 2023
Change-Id: I9a0988746eece3de0d61a11699cd11a54d8ec8b1 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4614288 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/11.5@{v8#17} Cr-Branched-From: 0c4044b-refs/heads/11.5.150@{#1} Cr-Branched-From: b71d303-refs/heads/main@{#87781}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
Jul 12, 2023
Change-Id: I17139cb1cd223eb8940e759c5d2fd32de720f145 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4678060 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/11.6@{v8#17} Cr-Branched-From: e29c028-refs/heads/11.6.189@{v8#3} Cr-Branched-From: 95cbef2-refs/heads/main@{#88340}
reposynch Bot
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
Sep 3, 2023
Change-Id: Ied9253f470fd32e5a0e4b6be2ec70396df259b80 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4815814 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/11.7@{v8#17} Cr-Branched-From: fe60869-refs/heads/11.7.439@{#1} Cr-Branched-From: aeb4552-refs/heads/main@{#89415}
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Performance and stability improvements on all platforms.
TBR=[email protected]
Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng
Change-Id: I93c472d3b286a2ae7e2694880ba2f6f0d77ebc83
Reviewed-on: https://chromium-review.googlesource.com/614983
Reviewed-by: v8 autoroll [email protected]
Cr-Commit-Position: refs/heads/6.2.238@{#1}
Cr-Branched-From: 13a990a-refs/heads/master@{#47351}