Conversation
Contributor
|
Thanks for your contribution! As this repo's description indicates, it is a mirror of V8's actual repository, so it's not possible to accept pull requests here. Please see https://v8.dev/docs/contribute for how to contribute to V8. |
Contributor
|
These changes landed: https://chromium.googlesource.com/v8/v8.git/+/180f017f50b63ce844ecb8d256c0717c4b9cf9d6 PR can be closed. |
billti
pushed a commit
to billti/v8
that referenced
this pull request
Apr 6, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: Ie689b6b301cb144e3b815af05b02fbb61069c6d5 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2071005 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/8.1@{v8#34} Cr-Branched-From: a4dcd39-refs/heads/8.1.307@{#1} Cr-Branched-From: f22c213-refs/heads/master@{#66031}
ghost
referenced
this pull request
in riscv-collab/v8
Jul 15, 2020
ghost
referenced
this pull request
in riscv-collab/v8
Jul 15, 2020
Resolve "test-run-machops/RunSpillLotsOfThings failed" Closes #34 See merge request swlab/swe/v8!42
palmer-dabbelt
pushed a commit
to palmer-dabbelt/v8
that referenced
this pull request
Jul 19, 2020
mulw sign extends its inputs
Kwizatz
pushed a commit
to AeonGames/v8
that referenced
this pull request
Sep 2, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: I7ee33fdecad1cff1797ece646e9ed1360f74a9b5 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2329619 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/8.5@{v8#34} Cr-Branched-From: a7f8bc4-refs/heads/8.5.210@{#1} Cr-Branched-From: dd58472-refs/heads/master@{#68510}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: I328e4b1705dd25067bb50450a96602fbe3791ec7 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1847839 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/7.8@{v8#34} Cr-Branched-From: 73694fd-refs/heads/7.8.279@{#1} Cr-Branched-From: 2314928-refs/heads/master@{#63555}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: I904624ffe7f0f8f09f88a9e0a49d61036c44410e Reviewed-on: https://chromium-review.googlesource.com/c/1398683 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/7.2@{v8#34} Cr-Branched-From: 6acd03c-refs/heads/7.2.502@{#1} Cr-Branched-From: b03041d-refs/heads/master@{#57910}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: I0d1979e1675dd29d84a247bbcd3f04eb43de2cc0 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1548592 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/7.4@{v8#34} Cr-Branched-From: 3e8a733-refs/heads/7.4.288@{#1} Cr-Branched-From: d077f9b-refs/heads/master@{#60039}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: I12bbc88874372c776cb250bda47f5fec0327779f Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1897618 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/7.9@{v8#34} Cr-Branched-From: be181e2-refs/heads/7.9.317@{#1} Cr-Branched-From: 0d7889d-refs/heads/master@{#64307}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
Merged: [ptr-compr, heap] Use system pointer size for heap limit computation Revision: 8732596 Merged: [ptr-compr, heap] Fix --huge_max_old_generation_size Revision: 23c09ad BUG=chromium:1045034,chromium:1049816 NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: I4a548324d2bc9864a5bd9826e8103ec7c874f1d2 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2050396 Reviewed-by: Igor Sheludko <[email protected]> Cr-Commit-Position: refs/branch-heads/8.0@{v8#34} Cr-Branched-From: 69827db-refs/heads/8.0.426@{v8#2} Cr-Branched-From: 2fe1552-refs/heads/master@{#65318}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
…g a function body Revision: 486ec80 [email protected] BUG=chromium:930580 LOG=N NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true Change-Id: I9aa6393b97afed7687e8481e2615dfc6f537a05b Reviewed-on: https://chromium-review.googlesource.com/c/1494757 Reviewed-by: Leszek Swirski <[email protected]> Cr-Commit-Position: refs/branch-heads/7.3@{v8#34} Cr-Branched-From: 9df9418-refs/heads/7.3.492@{v8#2} Cr-Branched-From: be213cf-refs/heads/master@{#59024}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
Revision: b8a9113 [email protected] BUG=chromium:898785 LOG=N NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true Change-Id: I6e7dcbb9235ddea19408ebae150a1e9598cd7414 Reviewed-on: https://chromium-review.googlesource.com/c/1317824 Reviewed-by: Benedikt Meurer <[email protected]> Commit-Queue: Benedikt Meurer <[email protected]> Cr-Commit-Position: refs/branch-heads/7.1@{v8#34} Cr-Branched-From: f70aaa8-refs/heads/7.1.302@{#1} Cr-Branched-From: 1dbcc78-refs/heads/master@{#56462}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
Large regexp results may exceed kMaxRegularHeapObjectSize and must thus be allocated in large object space. Drive-by: Rename '%InNewSpace' to '%InYoungGeneration'. (cherry picked from commit 4c15693) Tbr: [email protected] No-Try: true No-Presubmit: true No-Treechecks: true Bug: chromium:976627 Change-Id: I38b5aecb95a95cf2fdbb24d19550cec34361a09d Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1684176 Reviewed-by: Jakob Gruber <[email protected]> Commit-Queue: Jakob Gruber <[email protected]> Cr-Commit-Position: refs/branch-heads/7.6@{v8#34} Cr-Branched-From: 2cb2573-refs/heads/7.6.303@{#1} Cr-Branched-From: 201c509-refs/heads/master@{#61902}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
…t fields. Revision: f0e054c BUG=chromium:962588,v8:9233 LOG=N NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: I846353285132ebb17928d7e9e483eb43b3813f53 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1617243 Reviewed-by: Benedikt Meurer <[email protected]> Cr-Commit-Position: refs/branch-heads/7.5@{v8#34} Cr-Branched-From: 35b9bf5-refs/heads/7.5.288@{#1} Cr-Branched-From: 912b391-refs/heads/master@{#60911}
ceejatec
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Sep 24, 2020
Revision: 32b685f BUG=chromium:1100009 NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: Ia9af489c96e8ca1ca3e0c7f94ad812c165805b16 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2274889 Reviewed-by: Michael Achenbach <[email protected]> Commit-Queue: Liviu Rau <[email protected]> Cr-Commit-Position: refs/branch-heads/8.4@{v8#34} Cr-Branched-From: 88ed2e3-refs/heads/8.4.371@{#1} Cr-Branched-From: 35f88bf-refs/heads/master@{#67773}
Kwizatz
pushed a commit
to AeonGames/v8
that referenced
this pull request
Jul 25, 2021
We were inconsistent in handling offsets >= 2GB on 32-bit systems. The
code was still relying on this being detected as statically out of
bounds, but with the increase of {kV8MaxWasmMemoryPages} to support 4GB
memories, this is not the case any more.
This CL fixes this by again detecting such situations as statically OOB.
We do not expect to be able to allocate memories of size >2GB on such
systems. If this assumptions turns out to be wrong, we will erroneously
trap. If that happens, we will have to explicitly disallow memories of
such size on 32-bit systems.
Tbr: [email protected]
(cherry picked from commit 7ad5b96)
Bug: v8:7881, chromium:1201340
No-Try: true
No-Presubmit: true
No-Tree-Checks: true
Change-Id: Ib3d32b8d303eb047eb7811a045a8fa2b4ecb8cda
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2853596
Reviewed-by: Clemens Backes <[email protected]>
Commit-Queue: Clemens Backes <[email protected]>
Cr-Commit-Position: refs/branch-heads/9.1@{v8#34}
Cr-Branched-From: 0e4ac64-refs/heads/9.1.269@{#1}
Cr-Branched-From: f565e72-refs/heads/master@{#73847}
BenPope
pushed a commit
to BenPope/v8
that referenced
this pull request
Dec 10, 2021
Raise an exception if CallFunction is invoked with a class constructor. The check was initially removed in [1] but there are cases where we could end in CallFunction with class constructors from optimized code. [1] https://crrev.com/c/3186434 (cherry picked from commit 3e40f8f) Bug: chromium:1271384 Change-Id: I0d700c4b1d117334c1c4c14719e24cd1f2c5e3a2 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3291313 Commit-Queue: Patrick Thier <[email protected]> Reviewed-by: Jakob Gruber <[email protected]> Cr-Original-Commit-Position: refs/heads/main@{#77967} Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3306485 Cr-Commit-Position: refs/branch-heads/9.6@{v8#34} Cr-Branched-From: 0b7bda0-refs/heads/9.6.180@{#1} Cr-Branched-From: 41a5a24-refs/heads/main@{#77244}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
…n checks Revision: df98901 BUG=chromium:1171954 NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: Ib62b5cecb1f0c5a31856a26d97464e1ca941bc39 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2689191 Reviewed-by: Ross McIlroy <[email protected]> Cr-Commit-Position: refs/branch-heads/8.8@{v8#34} Cr-Branched-From: 2dbcdc1-refs/heads/8.8.278@{#1} Cr-Branched-From: 366d30c-refs/heads/master@{#71094}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: Id0ec1a120628ce12686218ef624cc259d2493b0d Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3020869 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/9.2@{v8#34} Cr-Branched-From: 5123834-refs/heads/9.2.230@{#1} Cr-Branched-From: 587a04f-refs/heads/master@{#74656}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: I64fbf0c521e571dfd6e51b18ecb0b576e01e902c Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2675184 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/8.9@{v8#34} Cr-Branched-From: 16b9bbb-refs/heads/8.9.255@{#1} Cr-Branched-From: d16a2a6-refs/heads/master@{#72039}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com Change-Id: Ibdf54931bb94c40238798966f182f574ba76466c Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3160106 Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/9.4@{v8#34} Cr-Branched-From: 3b51863-refs/heads/9.4.146@{#1} Cr-Branched-From: 2890419-refs/heads/master@{#76233}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
Change-Id: Id5ad771fb42b63cee9df610b7e5099875c6e05c4 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3748508 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/10.3@{v8#34} Cr-Branched-From: 1a8f4ca-refs/heads/10.3.174@{#1} Cr-Branched-From: 8fbefa4-refs/heads/main@{#80471}
cb-robot
pushed a commit
to couchbasedeps/v8-mirror
that referenced
this pull request
Aug 24, 2022
Merged: [compiler] Use MakeRefAssumeMemoryFence around closure initial map Revision: 9ef9e00 Merged: [compiler] fix 2 incorrect MakeRef usages in js-heap-broker.cc Revision: 7dc0a54 Merged: [compiler] Fix invalid MakeRef use in JSArrayRef::length_unsafe() Revision: cc6d54c Merged: [compiler] Fix invalid MakeRef uses in context specialization Revision: ee1e1fa Merged: [compiler] Fix invalid MakeRef uses in MapRef Revision: bb93f26 BUG=v8:12282,v8:7790 NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: Ib5ad274fc772a4a7e30619b4f893dce556209807 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3207891 Reviewed-by: Jakob Gruber <[email protected]> Cr-Commit-Position: refs/branch-heads/9.5@{v8#34} Cr-Branched-From: 4a03d61-refs/heads/9.5.172@{#1} Cr-Branched-From: 9a60704-refs/heads/main@{#76741}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
[email protected] Cq-Include-Trybots: luci.chromium.try:linux_chromium_rel_ng Change-Id: I84443545827c2ee8ddbc167e9b43b7f2699f63d2 Reviewed-on: https://chromium-review.googlesource.com/1183946 Reviewed-by: V8 Autoroller <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/6.9@{v8#34} Cr-Branched-From: d7b61ab-refs/heads/6.9.427@{#1} Cr-Branched-From: b7e108d-refs/heads/master@{#54504}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
Change-Id: I9a390f35f09eb9d9416b19fe805dca9be8d13981 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4568528 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/11.4@{v8#34} Cr-Branched-From: 8a8a1e7-refs/heads/11.4.183@{#1} Cr-Branched-From: 5483d8e-refs/heads/main@{#87241}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
May 31, 2023
Fixed: chromium:1378239 (cherry picked from commit db83e72) Change-Id: I5cbf9971ea34a58832bddc01a46cf47059cb5cd5 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3982253 Reviewed-by: Tobias Tebbi <[email protected]> Commit-Queue: Igor Sheludko <[email protected]> Cr-Commit-Position: refs/branch-heads/10.7@{v8#34} Cr-Branched-From: 4d2145c-refs/heads/10.7.193@{#1} Cr-Branched-From: 9521696-refs/heads/main@{#83201}
patrik-lengweiler
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
Jul 18, 2023
Change-Id: I0b458cd70612225ff35709aa3b5f56e66b8da2bb Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4689477 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/11.5@{v8#34} Cr-Branched-From: 0c4044b-refs/heads/11.5.150@{#1} Cr-Branched-From: b71d303-refs/heads/main@{#87781}
reposynch Bot
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
Sep 3, 2023
…t holey Bug: chromium:1473247 (cherry picked from commit ae7dc61) Change-Id: I5268513bc91ca0cc18e3e2115244c0b090afa0da Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4831892 Auto-Submit: Leszek Swirski <[email protected]> Owners-Override: Leszek Swirski <[email protected]> Commit-Queue: Darius Mercadier <[email protected]> Reviewed-by: Darius Mercadier <[email protected]> Commit-Queue: Leszek Swirski <[email protected]> Cr-Commit-Position: refs/branch-heads/11.6@{v8#34} Cr-Branched-From: e29c028-refs/heads/11.6.189@{v8#3} Cr-Branched-From: 95cbef2-refs/heads/main@{#88340}
reposynch Bot
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
Sep 24, 2023
The comment was correct, but the logic was wrong. In the example, the array is transitioned to dictionary mode, and in the next iteration, we crash when reading the next element. Fixed: chromium:1481179 Bug: v8:7700 (cherry picked from commit e7305f2) Change-Id: Ic1620dd54b01cae985658af361c32cd2f6a71280 No-try: true Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4882541 Auto-Submit: Victor Gomes <[email protected]> Reviewed-by: Leszek Swirski <[email protected]> Reviewed-by: Darius Mercadier <[email protected]> Commit-Queue: Darius Mercadier <[email protected]> Cr-Commit-Position: refs/branch-heads/11.7@{v8#34} Cr-Branched-From: fe60869-refs/heads/11.7.439@{#1} Cr-Branched-From: aeb4552-refs/heads/main@{#89415}
reposynch Bot
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
Nov 21, 2023
Merged: [maglev] Fix phi untagging skipping smi check Revision: 9e0df0b Merged: [maglev] Phi untagging fix Revision: 1a57a37 Merged: [maglev] Fix another phi untagging skipping smi check Revision: 5e88842 BUG=chromium:1499133,v8:7700 NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Change-Id: Ica7d399bd87f1b0d9308230f05d2cea9518d9bad Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5046675 Reviewed-by: Darius Mercadier <[email protected]> Cr-Commit-Position: refs/branch-heads/11.8@{v8#34} Cr-Branched-From: 935bdbf-refs/heads/11.8.172@{#1} Cr-Branched-From: b82a911-refs/heads/main@{#89779}
reposynch Bot
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
Dec 24, 2023
Change-Id: I08bd84d91ac1d4975557cbd6b5eb0b073723fb31 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5131898 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/12.1@{v8#34} Cr-Branched-From: b74ef6f-refs/heads/12.1.285@{#1} Cr-Branched-From: 32857fb-refs/heads/main@{#91313}
reposynch Bot
pushed a commit
to hexagon-geo-surv/v8
that referenced
this pull request
Jan 21, 2024
Since the work-around in depot_tools is still blocked, this just forces Python version 3.8 from V8's .vpython3 file when cpplint is used. (cherry picked from commit 3f32f3e) Bug: chromium:1517826 Change-Id: Ic8886efb800fe8f36711279bd7963799d471dc33 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5192365 Reviewed-by: Liviu Rau <[email protected]> Commit-Queue: Liviu Rau <[email protected]> Auto-Submit: Michael Achenbach <[email protected]> Cr-Original-Commit-Position: refs/heads/main@{#91803} Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5204833 Commit-Queue: Michael Achenbach <[email protected]> Reviewed-by: Eva Herencsárová <[email protected]> Cr-Commit-Position: refs/branch-heads/12.0@{v8#34} Cr-Branched-From: ed7b4ca-refs/heads/12.0.267@{#1} Cr-Branched-From: 210e75b-refs/heads/main@{#90651}
hubot
pushed a commit
that referenced
this pull request
Feb 19, 2024
Merged: [loong64][mips64][tagged-ptr] Make HeapNumber & BigInt layout classes Revision: c878c68 Merged: [loong64][sandbox] Port some sandbox patches to loong64 Revision: a04b0c4 NOTRY=true NOPRESUBMIT=true NOTREECHECKS=true [email protected] Bug: v8:14628 Change-Id: I7e2d1af6dd891e48d751441e53b4c5e29ec69624 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5303176 Auto-Submit: Zhao Jiazhong <[email protected]> Commit-Queue: Lutz Vahl <[email protected]> Reviewed-by: Lutz Vahl <[email protected]> Cr-Commit-Position: refs/branch-heads/12.2@{#34} Cr-Branched-From: 6eb5a96-refs/heads/12.2.281@{#1} Cr-Branched-From: 44cf56d-refs/heads/main@{#91934}
hubot
pushed a commit
that referenced
this pull request
Apr 25, 2024
Bug: 336961698 Change-Id: Ieb14876f5ea82dde55308c378b4ae40ec7df8c28 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5487859 Reviewed-by: Michael Achenbach <[email protected]> Commit-Queue: Alexander Schulze <[email protected]> Cr-Commit-Position: refs/branch-heads/12.3@{#34} Cr-Branched-From: a86e197-refs/heads/12.3.219@{#1} Cr-Branched-From: 21869f7-refs/heads/main@{#92385}
hubot
pushed a commit
that referenced
this pull request
May 16, 2024
Bug: 338908243 (cherry picked from commit cc05792) Change-Id: I9b5c2333924a54169ea3fa48e67e7db2ec67f6b9 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5545380 Reviewed-by: Jakob Kummerow <[email protected]> Commit-Queue: Matthias Liedtke <[email protected]> Auto-Submit: Matthias Liedtke <[email protected]> Commit-Queue: Jakob Kummerow <[email protected]> Cr-Commit-Position: refs/branch-heads/12.4@{#34} Cr-Branched-From: 309640d-refs/heads/12.4.254@{#1} Cr-Branched-From: 5dc2470-refs/heads/main@{#92862}
hubot
pushed a commit
that referenced
this pull request
Jun 17, 2024
So far the State struct in frames.cc stored the callee_pc_address, and when the callee_pc was read during frame iteration, it was read just in time from the callee_pc_address. This approach causes problems on arm64 because there the PC is authenticated, at least most of the time. However, fast API calls store their PC (the return address) on the isolate and not on the stack, and the PC on the isolate is not authenticated. Therefore, during frame iteration, a PC loaded from the isolate has to be handled differently than a PC loaded from the stack. At the code locations where the callee_pc is needed it is hard to tell if the callee_pc_address is on the isolate or on the stack. However, where the callee_pc_address gets initialized, it is easy. Since the callee_pc_address is not needed anyways, with this CL the callee_pc_address does not get stored anymore in the State, but instead the callee_pc directly. Bug: 341382353 (cherry picked from commit fee4ab5) Change-Id: I9c6441f6cf8490dbb5e0dc8bf18aa85e0375ea71 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5636650 Commit-Queue: Andreas Haas <[email protected]> Reviewed-by: Igor Sheludko <[email protected]> Cr-Commit-Position: refs/branch-heads/12.6@{#34} Cr-Branched-From: 3c9fa12-refs/heads/12.6.228@{#2} Cr-Branched-From: 981bb15-refs/heads/main@{#93835}
hubot
pushed a commit
that referenced
this pull request
Aug 5, 2024
Bug: 352690885 (cherry picked from commit af7f8f0) Change-Id: I8f7d2fb62a39117025be218fcc17dc74614f4e4a Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5762852 Commit-Queue: Rezvan Mahdavi Hezaveh <[email protected]> Reviewed-by: Rezvan Mahdavi Hezaveh <[email protected]> Cr-Commit-Position: refs/branch-heads/12.7@{#34} Cr-Branched-From: 35cc908-refs/heads/12.7.224@{#1} Cr-Branched-From: 6d60e67-refs/heads/main@{#94324}
hubot
pushed a commit
that referenced
this pull request
Aug 20, 2024
This avoids having to load the value back into a register if it was spilled inside of the loop. [email protected] (cherry picked from commit 9797576) Bug: chromium:360700873 Change-Id: I91f1e3c736547d25ec8aa8721c32105d49139309 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5800569 Commit-Queue: Jakob Kummerow <[email protected]> Reviewed-by: Jakob Kummerow <[email protected]> Cr-Commit-Position: refs/branch-heads/12.8@{#34} Cr-Branched-From: 70cbb39-refs/heads/12.8.374@{#1} Cr-Branched-From: 451b63e-refs/heads/main@{#95151}
hubot
pushed a commit
that referenced
this pull request
Sep 12, 2024
Change-Id: Id9f29ff4415a31b76e190299c3c31cf126a6ba7d Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5855029 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/12.9@{#34} Cr-Branched-From: 64a21d7-refs/heads/12.9.202@{#1} Cr-Branched-From: da4200b-refs/heads/main@{#95679}
hubot
pushed a commit
that referenced
this pull request
Oct 16, 2024
Version incremented at https://cr-buildbucket.appspot.com/build/8733910910929045889 Change-Id: Ibbd679552ed65e61b1132a7ea82f28bba2476834 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5937331 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/13.0@{#34} Cr-Branched-From: 4be854b-refs/heads/13.0.245@{#1} Cr-Branched-From: 1f5183f-refs/heads/main@{#96103}
hubot
pushed a commit
that referenced
this pull request
Dec 10, 2024
In addition to the field types we should also check the mutability. [email protected] Fixed: 382291459 (cherry picked from commit 3852cf8) Change-Id: I382164ed5534dd23c0c7a630690b57ae6ad7a4e1 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/6086196 Reviewed-by: Matthias Liedtke <[email protected]> Commit-Queue: Matthias Liedtke <[email protected]> Commit-Queue: Jakob Kummerow <[email protected]> Auto-Submit: Jakob Kummerow <[email protected]> Cr-Commit-Position: refs/branch-heads/13.2@{#34} Cr-Branched-From: 24068c5-refs/heads/13.2.152@{#1} Cr-Branched-From: 6054ba9-refs/heads/main@{#97085}
hubot
pushed a commit
that referenced
this pull request
Dec 11, 2024
When we decide that a loop phi should remain tagged, we call EnsurePhiInputsTagged to ensures that it only has tagged inputs, which calls EnsurePhiTagged, which might cause retagging of any untagged phi it has as input. In order to avoid retagging multiple times the same Phi, we have a SnaphotTable (`phi_taggings_`), which records existing tagging in the predecessors, and in which EnsurePhiTagged looks to avoid creating new retagging nodes. For loop phis, the backedge predecessor won't have an entry yet in this SnapshotTable (since we only visit loops once, this has to be the first time we visit the header and thus we can't have already visited the backedge block), and we should thus not call EnsurePhiTagged on the backedge. Note that the backedge input will anyways be properly tagged when FixLoopPhisBackedge is later called from the JumpLoop backedge. Fixed: chromium:382190919 (cherry picked from commit e4ecfc9) Change-Id: I7d4f05fa9cf61007f513f45f8763d8c74eda7412 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/6087795 Auto-Submit: Olivier Flückiger <[email protected]> Commit-Queue: Olivier Flückiger <[email protected]> Reviewed-by: Camillo Bruni <[email protected]> Commit-Queue: Camillo Bruni <[email protected]> Cr-Commit-Position: refs/branch-heads/13.1@{#34} Cr-Branched-From: 7998da6-refs/heads/13.1.201@{#1} Cr-Branched-From: 5e9af2a-refs/heads/main@{#96554}
hubot
pushed a commit
that referenced
this pull request
Jan 30, 2025
…n stability Bug: 390465670 (cherry picked from commit f920b6b) Change-Id: If6416403d1c3b23358050ea39791fef666d96111 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/6218883 Reviewed-by: Darius Mercadier <[email protected]> Commit-Queue: Darius Mercadier <[email protected]> Auto-Submit: Olivier Flückiger <[email protected]> Commit-Queue: Olivier Flückiger <[email protected]> Cr-Commit-Position: refs/branch-heads/13.3@{#34} Cr-Branched-From: 41dacff-refs/heads/13.3.415@{#1} Cr-Branched-From: 3348638-refs/heads/main@{#97937}
hubot
pushed a commit
that referenced
this pull request
Feb 28, 2025
Version incremented at https://cr-buildbucket.appspot.com/build/8721674469761925473 Change-Id: Ib61ba1015a55bde6258f7a25598769efa2e55cf4 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/6311710 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/13.4@{#34} Cr-Branched-From: 0f87a54-refs/heads/13.4.114@{#1} Cr-Branched-From: 27af2e9-refs/heads/main@{#98459}
hubot
pushed a commit
that referenced
this pull request
Jun 16, 2025
Version incremented at https://cr-buildbucket.appspot.com/build/8711910569704105633 Change-Id: I3d8574d60703c16b516c8235e29875f5e5a79de2 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/6646768 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/13.8@{#34} Cr-Branched-From: 61ddd47-refs/heads/13.8.258@{#1} Cr-Branched-From: fdb5de2-refs/heads/main@{#100480}
hubot
pushed a commit
that referenced
this pull request
Jun 17, 2025
Other Intl object address the same issue in https://issues.chromium.org/issues/42200234 Fixed: 412149700 (cherry picked from commit 1d82eec) Change-Id: I5af32a68c05d37925c0d4538822559c1493863e7 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/6652493 Owners-Override: Srinivas Sista <[email protected]> Commit-Queue: Frank Tang <[email protected]> Reviewed-by: Francis McCabe <[email protected]> Cr-Commit-Position: refs/branch-heads/13.7@{#34} Cr-Branched-From: dd5370d-refs/heads/13.7.152@{#1} Cr-Branched-From: fa9b753-refs/heads/main@{#99927}
hubot
pushed a commit
that referenced
this pull request
Aug 1, 2025
The embedder could pass a byte vector which points inside the sandbox, hence we should make sure to only read it once to protect against concurrent modification. This CL feeds the decoder the bytes from the copy we did anyways instead of using the incoming bytes. [email protected] (cherry picked from commit 1eda930) Bug: 433533359 Change-Id: I3767cb824a702f24c606730d4931fd8fdd8ac906 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/6811146 Commit-Queue: Jakob Kummerow <[email protected]> Auto-Submit: Clemens Backes <[email protected]> Reviewed-by: Jakob Kummerow <[email protected]> Cr-Commit-Position: refs/branch-heads/13.9@{#34} Cr-Branched-From: 76ea409-refs/heads/13.9.205@{#1} Cr-Branched-From: 2824212-refs/heads/main@{#100941}
hubot
pushed a commit
that referenced
this pull request
Nov 3, 2025
Version incremented at https://cr-buildbucket.appspot.com/build/8699238131104039537 Change-Id: I46850eff5f15e0f6b686e7dbd78073cb63befb8d Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/7108919 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/14.2@{#34} Cr-Branched-From: 37f82db-refs/heads/14.2.231@{#1} Cr-Branched-From: d1a6089-refs/heads/main@{#102804}
hubot
pushed a commit
that referenced
this pull request
Jan 9, 2026
…ive_gc_includes_global Original change's description: > [heap] Guard MaximumGlobalMemorySizeFromV8Size behind ineffective_gc_includes_global > > MaximumGlobalMemorySizeFromV8Size is orthogonal > to external_memory_accounted_in_global_limit, and it likely > doesn't help results; it should go along ineffective_gc_includes_global > to be experimented separately. > > Bug: 361124432 > Change-Id: Ib267aa64f6bed0259b2dd048afbaf44afd6cb947 > Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/7379666 > Commit-Queue: Etienne Pierre-Doray <[email protected]> > Reviewed-by: Michael Lippautz <[email protected]> > Cr-Commit-Position: refs/heads/main@{#104516} (cherry picked from commit e51fdff) Bug: 474315655,361124432 Change-Id: Ib267aa64f6bed0259b2dd048afbaf44afd6cb947 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/7416151 Auto-Submit: Chrome Cherry Picker <chrome-cherry-picker@chops-service-accounts.iam.gserviceaccount.com> Bot-Commit: Rubber Stamper <[email protected]> Commit-Queue: Rubber Stamper <[email protected]> Cr-Commit-Position: refs/branch-heads/14.4@{#34} Cr-Branched-From: 80acc26-refs/heads/14.4.258@{#1} Cr-Branched-From: ce7e597-refs/heads/main@{#104020}
hubot
pushed a commit
that referenced
this pull request
Jan 13, 2026
…Pool" This reverts commit 30c0047. Reason for revert: b/462213735 Original change's description: > [arm64] Use CHECK (instead of DCHECK) in CheckVeennerPool > > This is a hardening measure to prevent potential data corruption > by crashing immediately if the invariant is ever violated in release > mode. > > Bug: 435640828 > Change-Id: Idc10542a68d52747aa1e8500ae3ef3a2cf138330 > Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/7014092 > Auto-Submit: Victor Gomes <[email protected]> > Commit-Queue: Victor Gomes <[email protected]> > Reviewed-by: Anton Bikineev <[email protected]> > Cr-Commit-Position: refs/heads/main@{#102961} Bug: 462213735 Fixed: 475265929 (cherry picked from commit ff77083) Change-Id: I6078b69e87243a1bfbee279ffa42bb3abe8a534c Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/7452031 Reviewed-by: Deepti Gandluri <[email protected]> Commit-Queue: Rezvan Mahdavi Hezaveh <[email protected]> Cr-Commit-Position: refs/branch-heads/14.3@{#34} Cr-Branched-From: 13c7e31-refs/heads/14.3.127@{#1} Cr-Branched-From: 01af089-refs/heads/main@{#103352}
hubot
pushed a commit
that referenced
this pull request
Mar 6, 2026
Version incremented at https://cr-buildbucket.appspot.com/build/8688076396695693537 Change-Id: I3a44e6cea4552ae39b06bf4e0f1e2312fe9f0558 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/7638458 Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com> Cr-Commit-Position: refs/branch-heads/14.6@{#34} Cr-Branched-From: e04c3a1-refs/heads/14.6.202@{#1} Cr-Branched-From: 3b0b01e-refs/heads/main@{#105132}
hubot
pushed a commit
that referenced
this pull request
Apr 7, 2026
…sm frames The equality operator for FrameStateFunctionInfo did not compare the wasm signature field in JSToWasmFrameStateFunctionInfo. This could cause CSE to incorrectly merge FrameState nodes with different wasm signatures, leading to the deoptimizer using the wrong return type when materializing a JS-to-Wasm builtin continuation frame. (cherry picked from commit 0739828) Bug: 497404188 Change-Id: I671cda5784089dd9875d90c5f48e8580cb5fa697 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/7709449 Reviewed-by: Matthias Liedtke <[email protected]> Reviewed-by: Daniel Lehmann <[email protected]> Commit-Queue: Matthias Liedtke <[email protected]> Cr-Original-Commit-Position: refs/heads/main@{#106175} Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/7735261 Cr-Commit-Position: refs/branch-heads/14.7@{#34} Cr-Branched-From: 723547b-refs/heads/14.7.173@{#1} Cr-Branched-From: 3fc49d4-refs/heads/main@{#105661}
hubot
pushed a commit
that referenced
this pull request
May 12, 2026
…isions) https://chromium.googlesource.com/external/github.com/WebKit/JetStream.git/+log/2145cedef4ca..91552391bf2f 2026-05-06 [email protected] Fix wrong update-ui performance.mark and measure nesting 2026-03-26 [email protected] Improve memory script 2026-03-19 [email protected] Address comments for memory measurement script 2026-03-19 [email protected] Add memory measurement script 2026-03-18 [email protected] Actually fix --dump-test-list 2026-03-18 [email protected] Fix --dump-test-list and --test in v3.0-custom 2026-03-17 [email protected] Update v3.0-custom to latest v3.0 branch 2026-01-30 [email protected] Merge pull request #272 from danleh/JetStream3.0 2026-01-19 [email protected] Reland: Improve Benchmark constructor and instance variables #255 (#263) 2026-01-19 [email protected] Reland: Add resource loading bar (#264) 2026-01-19 [email protected] Merge pull request #261 from danleh/dotnet-mime-fix 2026-01-17 [email protected] Merge pull request #262 from kmiller68/revert-replace-prefetch-logic 2026-01-15 [email protected] Merge pull request #245 from kmiller68/add-open-governance-policy 2026-01-15 [email protected] Merge pull request #260 from danleh/readme-addendum 2026-01-14 [email protected] Merge pull request #258 from danleh/repo-readme 2026-01-14 [email protected] Improve Benchmark constructor and instance variables (#255) 2026-01-14 [email protected] Merge pull request #254 from kmiller68/scatter-plot-should-increase-with-score 2026-01-14 [email protected] Merge pull request #257 from danleh/remove-unused-deps 2026-01-14 [email protected] Fix dev server output (#252) 2026-01-14 [email protected] Merge pull request #256 from danleh/fix-npm-audit 2026-01-14 [email protected] Add resource loading bar (#246) 2026-01-13 [email protected] Replace prefetch logic (#251) 2026-01-06 [email protected] Merge pull request #249 from kmiller68/add-verbose-option-to-server 2026-01-06 [email protected] Add BrowserFileLoader (#242) 2025-12-18 [email protected] Add --force-gc support (#248) 2025-12-10 [email protected] Merge pull request #244 from kmiller68/nonStandard-error-rotate-back-to-red 2025-12-10 [email protected] fewer shell warnings (#234) 2025-11-13 [email protected] Merge pull request #238 from kmiller68/transformers-js-loads-resources-at-runtime 2025-11-13 [email protected] Merge pull request #239 from Constellation/mod-256 2025-11-12 [email protected] Merge pull request #225 from kmiller68/reduce-kotlin-iteration-count 2025-11-07 [email protected] Merge pull request #236 from Constellation/use-integer 2025-11-05 [email protected] sync-fs and async-fs improvements (#217) 2025-11-04 [email protected] More ubuntu tests (#235) 2025-11-04 [email protected] Merge pull request #188 from kmiller68/dotnet-legacy-exceptions 2025-10-30 [email protected] Github Workflow: split tests (#227) 2025-10-30 [email protected] Enable mobx-startup (#233) 2025-10-30 [email protected] More top-level cleanup (#229) 2025-10-30 [email protected] Improve preload code (#230) 2025-10-30 [email protected] Merge pull request #226 from kmiller68/console-warn-and-error-stderr-clis 2025-10-30 [email protected] Fix npm run test:build (#232) 2025-10-30 [email protected] Merge pull request #212 from kmiller68/disable-whisper-wasm 2025-10-29 [email protected] New Workload: Babylon class-heavy workload (#143) 2025-10-29 [email protected] Avoid new Date and improve filename extraction in doxbee benchmarks (#231) 2025-10-28 [email protected] New Workload: mobx observer + startup (#151) 2025-10-28 [email protected] StartupBenchmark refactoring (#218) 2025-10-27 [email protected] Add post-quit sleep for Safari (#220) 2025-10-27 [email protected] Enable jsdom-d3-startup by default. (#223) 2025-10-27 [email protected] More consistent LICENSE.txt files (#221) 2025-10-27 [email protected] Move more code to utils (#219) 2025-10-23 [email protected] Add npm run test:build script (#189) 2025-10-23 [email protected] Merge pull request #216 from kmiller68/total-to-overall-summary 2025-10-23 [email protected] render non default params (#197) 2025-10-23 [email protected] New Workload: JSDOM + d3 (#124) 2025-10-22 [email protected] Update test.yml (#190) 2025-10-22 [email protected] Await driver.quit (#209) 2025-10-22 [email protected] Add explicit version checks (#211) 2025-10-22 [email protected] Improve iterationCount and worstCaseCount params parsing (#215) 2025-10-22 [email protected] More consistent total score printing (#213) 2025-10-22 [email protected] Do not trigger logo animations again after loading all resources (#214) 2025-10-22 [email protected] support startAutomatically (#201) 2025-10-22 [email protected] Conditionally show more metric details (#167) 2025-10-22 [email protected] fix gaussian-blur input data (#207) 2025-10-22 [email protected] Improve tags (#206) 2025-10-22 [email protected] update selenium-webdriver version (#208) 2025-10-21 [email protected] Auto decompress files as part of test running (#195) 2025-10-17 [email protected] Support multiple comma separate tests in URL params (#198) 2025-10-15 [email protected] Merge pull request #77 from Constellation/base64 2025-10-15 [email protected] Fix prismjs sample file name(#202) 2025-10-08 [email protected] Add prettier config (#176) 2025-10-07 [email protected] Merge pull request #170 from eqrion/zlib-compress 2025-10-03 [email protected] Merge pull request #194 from victorvianna/main 2025-10-03 [email protected] Change shouldReport to report back 2025-10-03 [email protected] Compact display for non-default params (#191) 2025-10-01 [email protected] Add group-details parameter to conditionally display grouped benchmarks (#157) 2025-10-01 [email protected] New Workload: prismjs source code highlighting (#149) 2025-10-01 [email protected] New Workload: validatorjs regexp heavy workload (#147) 2025-09-23 [email protected] Merge pull request #192 from Constellation/dump-test-list 2025-09-22 [email protected] New Workload: TypeScript standalone (#117) 2025-09-19 [email protected] Merge pull request #187 from kmiller68/fix-web-ssr-for-options-bag 2025-09-17 [email protected] Test is causing jetsams on iOS. Disabling for now while I investigate. 2025-09-17 [email protected] Add JetStreamParams (#171) 2025-09-17 [email protected] New Workload: React Server Side Rendering Startup (#116) 2025-09-17 [email protected] Merge pull request #130 from danleh/j2cl-box2d 2025-09-16 [email protected] Merge pull request #182 from danleh/recompile-dotnet 2025-09-16 [email protected] Merge pull request #148 from danleh/transformersjs-rebase 2025-09-11 [email protected] Experimental Workload: Intl (#146) 2025-09-10 [email protected] Merge pull request #178 from danleh/disable-gcc-loops-quicksort 2025-09-08 [email protected] Merge pull request #168 from kmiller68/reduce-dotnet-runtime 2025-09-03 [email protected] More consistent metric reporting (#175) 2025-09-03 [email protected] Minor browser test improvements (#177) 2025-09-03 [email protected] Merge pull request #173 from danleh/text-decoder-polyfill-move 2025-09-03 [email protected] Remove unused scripts (#174) 2025-09-02 [email protected] Merge pull request #172 from danleh/compress-js 2025-09-01 [email protected] Update/rewrite web-tooling-benchmark (#153) 2025-09-01 [email protected] Support custom benchmark arguments (#165) 2025-09-01 [email protected] Status update and CSS fixes (#169) 2025-08-27 [email protected] Use event.key instead of event.which #166 2025-08-27 [email protected] Merge pull request #144 from danleh/logo-text-to-path 2025-08-26 [email protected] StartupBenchmark: add better assertion failures (#164) 2025-08-26 [email protected] Add UTF16 sources test (#162) 2025-08-26 [email protected] Convert startup-helper/BabelCacheBuster.js to esm (#163) 2025-08-26 [email protected] Add startup benchmark helper (#158) 2025-08-20 [email protected] Merge pull request #159 from danleh/dotnet-wasm-rename 2025-08-15 [email protected] Expose JetStream global to workloads (#139) 2025-08-15 [email protected] Merge pull request #145 from Constellation/tfjs 2025-08-14 [email protected] Only expose a console copy in the realm (#137) 2025-08-14 [email protected] Merge pull request #133 from kmiller68/add-grouped-benchmark-for-sunspider 2025-08-14 [email protected] Merge pull request #142 from danleh/Dart-remove-unused-runner 2025-08-14 [email protected] Merge pull request #140 from danleh/disable-HashSet-wasm 2025-08-14 [email protected] merge async-fs code (#135) 2025-08-14 [email protected] Enable UniPoker validation by default (#141) 2025-08-14 [email protected] Merge pull request #138 from danleh/kotlin-fix-isD8-access 2025-08-14 [email protected] Bump selenium-webdriver version (#136) 2025-08-14 [email protected] Merge pull request #97 from eqrion/dart-todomvc 2025-08-14 [email protected] Improve console (#131) 2025-08-14 [email protected] Improve runner code (#134) 2025-08-14 [email protected] Rename worst4 to just worst (#129) 2025-08-13 [email protected] Merge pull request #127 from Constellation/list 2025-08-13 [email protected] Merge pull request #84 from danleh/kotlin 2025-08-12 [email protected] CSS fixes for info button (#126) 2025-08-12 [email protected] Add info-button (#123) 2025-08-11 [email protected] Merge pull request #119 from Constellation/set-status 2025-08-11 [email protected] Display workload scatterplots (#120) 2025-08-11 [email protected] Update .gitignore (#121) 2025-08-07 [email protected] Merge pull request #115 from kmiller68/fix-bencharks-by-tag-testlist 2025-08-06 [email protected] Configurable resource prefetching (#70) 2025-08-06 [email protected] github workflow: keep running all tests even if there are failures (#114) 2025-08-05 [email protected] Add Scripts helper classes (#109) 2025-08-05 [email protected] Add three.js benchmark (#75) 2025-08-04 [email protected] Add validate method for async-fs and sync-fs (#111) 2025-08-04 [email protected] Add more performance.measure (#110) 2025-07-31 [email protected] Remove LuaJSFight micro benchmarks (#104) 2025-07-31 [email protected] Add cli flags (#101) 2025-07-31 [email protected] Cleanup performance.now shimming (#108) 2025-07-30 [email protected] Formatting (#105) 2025-07-30 [email protected] Run all disabled workloads in tests (#103) 2025-07-30 [email protected] Add benchmarks argument to Driver (#98) 2025-07-29 [email protected] Merge pull request #102 from kmiller68/reenable-dotnet-tests 2025-07-29 [email protected] Simplify updateUIAfterRun (#99) 2025-07-29 [email protected] Merge customTestList and testList (#100) 2025-07-29 [email protected] Use tags for workload selection (#82) 2025-07-23 [email protected] Merge pull request #93 from danleh/cleanup-loadInternal 2025-07-23 [email protected] Merge pull request #73 from maraf/DotnetWasm 2025-07-21 [email protected] Fix typo in headers fetch request (#80) 2025-07-15 [email protected] use default performance.measure implementation (#83) 2025-07-15 [email protected] Use subScores directly for the UI (#81) 2025-07-15 [email protected] Add basic unittests (#94) 2025-07-14 [email protected] Fix accidental loading status clipping in wide windows (#85) 2025-07-14 [email protected] Simplify formatting (#79) 2025-07-12 [email protected] Merge pull request #90 from kmiller68/fix-prepareForNextIteration-async-benchmark 2025-07-10 [email protected] Merge pull request #74 from kmiller68/preload-blobs-wasm 2025-07-08 [email protected] Merge pull request #76 from kmiller68/zlib-check-_compressFile-to-setupModule 2025-07-02 [email protected] Revert #78 as it broke JetStream3 on web page 2025-07-02 [email protected] Remove custom run methods #78 2025-06-04 [email protected] Merge pull request #72 from Constellation/new-string 2025-05-13 [email protected] Merge pull request #71 from kmiller68/disable-extra-bigint-tests 2025-05-13 [email protected] Merge pull request #65: OfflineAssembler: Avoid accidental toString operation 2025-05-13 [email protected] Merge pull request #64: WSL: Add performance.mark 2025-05-13 [email protected] Merge pull request #60: date-format: Add result hashing to prevent DCE 2025-05-13 [email protected] Merge pull request #62: Avoid eval in date-format-tofte 2025-05-05 [email protected] Merge pull request #53 from mkustermann/js-strings 2025-04-24 [email protected] Merge pull request #67 from issackjohn/issack/add-edge 2025-04-24 [email protected] Merge pull request #61 from camillobruni/2025-03-26_gbemu_hash 2025-04-07 [email protected] Merge pull request #59 from camillobruni/2025-03-26_cli_error_handling 2025-04-07 [email protected] Merge pull request #42 from camillobruni/2025-01-28_cli_tests 2025-04-07 [email protected] Merge pull request #41 from camillobruni/2025-01-28_runner_log 2025-04-07 [email protected] Merge pull request #14 from camillobruni/2024-11-26_tags 2025-03-12 [email protected] Merge pull request #57 from kmiller68/module-url-fix-v2 2025-02-27 [email protected] Merge pull request #51 from danleh/main 2025-02-27 [email protected] Merge pull request #52 from kmiller68/pathname-browser-dart 2025-02-26 [email protected] Merge pull request #33 from danleh/dart-flute-wasm-squash 2025-02-26 [email protected] Merge pull request #47 from kmiller68/argon2-to-js-scoring-v2 2025-02-26 [email protected] Merge pull request #40 from danleh/8bitbench-wasm-scoring 2025-02-12 [email protected] Merge pull request #43 from danleh/zlib-wasm 2025-02-10 [email protected] Merge pull request #45 from danleh/richards-workload-size-as-input 2025-02-07 [email protected] Merge pull request #44 from danleh/wasm-workload-sizes 2025-01-28 [email protected] Merge pull request #39 from camillobruni/2025-01-28_test_runner_fix 2025-01-27 [email protected] Merge pull request #35 from danleh/hashset-wasm-scoring 2025-01-27 [email protected] Merge pull request #30 from danleh/sqlite-scoring 2025-01-27 [email protected] Merge pull request #28 from danleh/quicksort-wasm-scoring 2025-01-27 [email protected] Merge pull request #37 from WebKit/remove-octane-zlib 2025-01-24 [email protected] Revert "Remove Octane-zlib." 2025-01-24 [email protected] Remove Octane-zlib. 2025-01-24 [email protected] Merge pull request #13 from camillobruni/2024-11-22_driver_prep 2025-01-24 [email protected] Merge pull request #11 from camillobruni/2024-11-26_github_actions 2025-01-23 [email protected] Merge pull request #27 from danleh/richards-js-scoring 2025-01-22 [email protected] Merge pull request #34 from WebKit/fix-report-JSON 2025-01-22 [email protected] Merge pull request #32 from danleh/performance-mark-measure 2025-01-21 [email protected] Merge pull request #31 from WebKit/revert-tfjs-to-js-scoring 2025-01-21 [email protected] Merge pull request #16 from camillobruni/2024-11-26_ms 2025-01-21 [email protected] Merge pull request #26 from danleh/gcc-loops-js-scoring 2025-01-16 [email protected] Merge pull request #23 from WebKit/tfjs-to-js-scoring 2025-01-16 [email protected] Merge pull request #17 from camillobruni/2024-11-26_safe_add 2025-01-16 [email protected] Merge pull request #25 from danleh/spidermonkey-cli-fix 2025-01-08 [email protected] Merge pull request #21 from WebKit/remove-float-mm-c 2024-12-21 [email protected] Merge pull request #18 from WebKit/wasm-benchmarks-should-be-scored-like-js-benchmarks 2024-12-20 [email protected] Merge pull request #2 from danleh/sqlite-wasm 2024-12-09 [email protected] Merge pull request #5 from camillobruni/2024-11-22_runner_improvements 2024-12-09 [email protected] Merge pull request #8 from camillobruni/2024-11-22_regexp 2024-12-09 [email protected] Merge pull request #19 from camillobruni/2024-11-27_async_fs 2024-11-26 [email protected] Merge pull request #7 from camillobruni/2024-11-22_driver_vars 2024-11-25 [email protected] Merge pull request #4 from camillobruni/2024-11-22_3dcube_fix 2024-11-25 [email protected] Merge pull request #6 from camillobruni/2024-11-22_gitignore 2024-11-13 [email protected] Merge pull request #1 from danleh/main 2024-10-30 [email protected] Initial copy from WebKit/PerformanceTests/JetStream3 2024-10-30 [email protected] Initial commit If this roll has caused a breakage, revert this CL and stop the roller using the controls here: https://autoroll.skia.org/r/jetstream-v3-0-custom-v8 Please CC [email protected],[email protected],[email protected] on the revert to ensure that a human is aware of the problem. To file a bug in JetStream v3.0-custom: https://issues.chromium.org/u/1/issues/new?component=1741045&pli=1&template=0 To file a bug in V8: https://issues.chromium.org/u/1/issues/new?component=1456824&pli=1&template=2109136 To report a problem with the AutoRoller itself, please file a bug: https://issues.skia.org/issues/new?component=1389291&template=1850622 Documentation for the AutoRoller is here: https://skia.googlesource.com/buildbot/+doc/main/autoroll/README.md Bug: None Tbr: [email protected],[email protected] Change-Id: Ic53c7e030a05111ba56e03ab158282762aa17360 Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/7823185 Commit-Queue: Camillo Bruni <[email protected]> Bot-Commit: [email protected] <[email protected]> Reviewed-by: Camillo Bruni <[email protected]> Cr-Commit-Position: refs/heads/main@{#107273}
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
https://portingguide.readthedocs.io/en/latest/builtins.html#removed-execfile
flake8 testing of https://github.com/v8/v8 on Python 3.7.1
$ flake8 . --count --select=E9,F63,F7,F82 --show-source --statistics
@rryk
@bmsdave