Skip to content

Version 6.8.188 - #25

Closed
MissSpazzyBitch wants to merge 1 commit into
masterfrom
6.8.188
Closed

MissSpazzyBitch wants to merge 1 commit into
masterfrom
6.8.188

Conversation

@MissSpazzyBitch

Copy link
Copy Markdown

Performance and stability improvements on all platforms.

TBR=[email protected]

Cq-Include-Trybots: luci.chromium.try:linux_chromium_rel_ng
Change-Id: Id0ec0565b3de0ce3e287b799662abe189eb2106a
Reviewed-on: https://chromium-review.googlesource.com/1048586
Reviewed-by: v8 autoroll [email protected]
Cr-Commit-Position: refs/heads/6.8.188@{#1}
Cr-Branched-From: 73933f2-refs/heads/master@{#53044}

Performance and stability improvements on all platforms.

[email protected]

Cq-Include-Trybots: luci.chromium.try:linux_chromium_rel_ng
Change-Id: Id0ec0565b3de0ce3e287b799662abe189eb2106a
Reviewed-on: https://chromium-review.googlesource.com/1048586
Reviewed-by: v8 autoroll <[email protected]>
Cr-Commit-Position: refs/heads/6.8.188@{#1}
Cr-Branched-From: 73933f2-refs/heads/master@{#53044}
@psmarshall psmarshall closed this May 8, 2018
dumganhar pushed a commit to dumganhar/v8_fork that referenced this pull request Jun 5, 2018
Revision: 6b31174

NOTRY=true
NOPRESUBMIT=true
NOTREECHECKS=true
[email protected]

Bug: chromium:726636
Change-Id: I415290ca2f3aaff9ec29b285a5883815205f4ddd
Reviewed-on: https://chromium-review.googlesource.com/529625
Reviewed-by: Adam Klein <[email protected]>
Cr-Commit-Position: refs/branch-heads/6.0@{v8#25}
Cr-Branched-From: 97dbf62-refs/heads/6.0.286@{#1}
Cr-Branched-From: 12e6f1c-refs/heads/master@{#45439}
hubot pushed a commit that referenced this pull request Jun 11, 2018
TBR=santa

Change-Id: I7670e1b62b91615ae7b588c6e41a4ac3890de2a4
Reviewed-on: https://chromium-review.googlesource.com/1094642
Reviewed-by: Michael Achenbach <[email protected]>
Cr-Commit-Position: refs/branch-heads/6.8@{#25}
Cr-Branched-From: 44d7d7d-refs/heads/6.8.275@{#1}
Cr-Branched-From: 5754f66-refs/heads/master@{#53286}
hubot pushed a commit that referenced this pull request Aug 7, 2018
Revision: 3656b46

BUG=chromium:869313
LOG=N
NOTRY=true
NOPRESUBMIT=true
NOTREECHECKS=true
[email protected]

Change-Id: Ia71f6318ee07356d74e1badb49c5c1402fb0815b
Reviewed-on: https://chromium-review.googlesource.com/1165230
Reviewed-by: Tobias Tebbi <[email protected]>
Commit-Queue: Tobias Tebbi <[email protected]>
Cr-Commit-Position: refs/branch-heads/6.9@{#25}
Cr-Branched-From: d7b61ab-refs/heads/6.9.427@{#1}
Cr-Branched-From: b7e108d-refs/heads/master@{#54504}
peterwmwong pushed a commit to peterwmwong/v8 that referenced this pull request Dec 9, 2018
This reverts commit caed2cc.

Reason for revert: Breaks layout tests, e.g.

https://test-results.appspot.com/data/layout_results/V8-Blink_Linux_64__dbg_/14924/webkit_layout_tests%20%28with%20patch%29/layout-test-results/results.html

crash log for renderer (pid <unknown>):
STDOUT: <empty>
STDERR: 
STDERR: 
STDERR: #
STDERR: # Fatal error in ../../v8/src/base/platform/elapsed-timer.h, line 24
STDERR: # Debug check failed: !IsStarted().
STDERR: #
STDERR: #
STDERR: #
STDERR: #FailureMessage Object: 0x7ffc46707640#0 0x565409263b6f base::debug::StackTrace::StackTrace()
STDERR: #1 0x56540a8a32fb gin::(anonymous namespace)::PrintStackTrace()
STDERR: #2 0x56540a8980d8 V8_Fatal()
STDERR: #3 0x56540a897e35 v8::base::(anonymous namespace)::DefaultDcheckHandler()
STDERR: v8#4 0x565407971f02 v8::base::ElapsedTimer::Start()
STDERR: v8#5 0x565407d08edf v8::internal::TimedHistogram::Start()
STDERR: v8#6 0x565407e500d5 v8::internal::IncrementalMarking::AdvanceIncrementalMarkingOnAllocation()
STDERR: v8#7 0x565407e4f977 v8::internal::IncrementalMarking::Observer::Step()
STDERR: v8#8 0x565407e48092 v8::internal::AllocationObserver::AllocationStep()
STDERR: v8#9 0x565407eb0751 v8::internal::SpaceWithLinearArea::InlineAllocationStep()
STDERR: v8#10 0x565407eb3e44 v8::internal::NewSpace::EnsureAllocation()
STDERR: v8#11 0x565407e258ff v8::internal::NewSpace::AllocateRaw()
STDERR: v8#12 0x565407e06b2d v8::internal::Heap::AllocateRaw()
STDERR: v8#13 0x565407e432ef v8::internal::Heap::AllocateRawWithLightRetry()
STDERR: v8#14 0x565407e433cf v8::internal::Heap::AllocateRawWithRetryOrFail()
STDERR: v8#15 0x565407e04d48 v8::internal::Factory::NewFixedArrayWithFiller()
STDERR: v8#16 0x565407fd6339 v8::internal::HashTable<>::New()
STDERR: v8#17 0x565407fd7be8 v8::internal::HashTable<>::EnsureCapacity()
STDERR: v8#18 0x565407fc7e95 v8::internal::Dictionary<>::Add()
STDERR: v8#19 0x565407fcf453 v8::internal::BaseNameDictionary<>::Add()
STDERR: v8#20 0x565407f89ee4 v8::internal::LookupIterator::ApplyTransitionToDataProperty()
STDERR: v8#21 0x5654080036e2 v8::internal::Object::AddDataProperty()
STDERR: v8#22 0x56540793061f v8::internal::(anonymous namespace)::DefineDataProperty()
STDERR: v8#23 0x56540792da59 v8::internal::(anonymous namespace)::InstantiateObject()
STDERR: v8#24 0x56540792b75a v8::internal::(anonymous namespace)::InstantiateFunction()
STDERR: v8#25 0x56540792b4db v8::internal::ApiNatives::InstantiateFunction()
STDERR: v8#26 0x5654079594bf v8::FunctionTemplate::GetFunction()
STDERR: v8#27 0x56540a7af74e blink::V8ObjectConstructor::CreateInterfaceObject()
STDERR: v8#28 0x56540a7afe01 blink::V8PerContextData::ConstructorForTypeSlowCase()
STDERR: v8#29 0x56540a7afdd6 blink::V8PerContextData::ConstructorForTypeSlowCase()
STDERR: v8#30 0x56540a7afdd6 blink::V8PerContextData::ConstructorForTypeSlowCase()
STDERR: v8#31 0x56540a7afcb4 blink::V8PerContextData::CreateWrapperFromCacheSlowCase()
STDERR: v8#32 0x56540a7aef73 blink::V8DOMWrapper::CreateWrapper()
STDERR: v8#33 0x56540a7abf6b blink::ScriptWrappable::Wrap()
STDERR: v8#34 0x56540a677199 blink::V8Document::documentElementAttributeGetterCallback()
STDERR: v8#35 0x565407a0aec3 v8::internal::FunctionCallbackArguments::Call()
STDERR: v8#36 0x565407a097be v8::internal::(anonymous namespace)::HandleApiCallHelper<>()
STDERR: v8#37 0x565407a0877b v8::internal::Builtins::InvokeApiFunction()
STDERR: v8#38 0x565407fe785a v8::internal::Object::GetPropertyWithAccessor()
STDERR: v8#39 0x565407fe697e v8::internal::Object::GetProperty()
STDERR: v8#40 0x565407ec8c71 v8::internal::LoadIC::Load()
STDERR: v8#41 0x565407ed6401 v8::internal::__RT_impl_Runtime_LoadIC_Miss()
STDERR: v8#42 0x5654087593f2 <unknown>
STDERR: [16162:16185:1122/143518.356897:WARNING:crash_handler_host_linux.cc(341)] Could not translate tid, attempt = 1 retry ...


Original change's description:
> [heap] Improve embedder tracing during incremental marking
> 
> Add a path into embedder tracing on allocation. This is safe as as Blink
> is not allowed to call into V8 during object construction.
> 
> Bug: chromium:843903
> Change-Id: I5af053c3169f5a33778ebce5d7c5c43e4efb1aa4
> Reviewed-on: https://chromium-review.googlesource.com/c/1348749
> Commit-Queue: Michael Lippautz <[email protected]>
> Reviewed-by: Ulan Degenbaev <[email protected]>
> Cr-Commit-Position: refs/heads/master@{#57757}

[email protected],[email protected]

Change-Id: Ide2c0b284b52bee17573adcc89f14be4e40dab91
No-Presubmit: true
No-Tree-Checks: true
No-Try: true
Bug: chromium:843903
Reviewed-on: https://chromium-review.googlesource.com/c/1349189
Reviewed-by: Yang Guo <[email protected]>
Commit-Queue: Yang Guo <[email protected]>
Cr-Commit-Position: refs/heads/master@{#57759}
junhuaw pushed a commit to junhuaw/v8 that referenced this pull request Feb 6, 2019
This makes sure we properly recognize a newline character as part of
semicolon insertion, even if the newline appears after a CPP-style
single line comment. The same applies for newlines within C-style multi
line comments.

[email protected]
TEST=mjsunit/asm/regress-913822
BUG=chromium:913822

Change-Id: Ib8c9d3d923ced2fb9a4d37ee906f38abd2ee5634
Originally-reviewed-on: https://chromium-review.googlesource.com/c/1373551
No-Try: true
No-Presubmit: true
No-Treechecks: true
Reviewed-on: https://chromium-review.googlesource.com/c/1382745
Reviewed-by: Andreas Haas <[email protected]>
Commit-Queue: Clemens Hammacher <[email protected]>
Cr-Commit-Position: refs/branch-heads/7.2@{v8#25}
Cr-Branched-From: 6acd03c-refs/heads/7.2.502@{#1}
Cr-Branched-From: b03041d-refs/heads/master@{#57910}
Teemperor pushed a commit to Teemperor/v8 that referenced this pull request May 25, 2019
Revision: c4eae87

BUG=chromium:942699
LOG=N
NOTRY=true
NOPRESUBMIT=true
NOTREECHECKS=true
[email protected]

Change-Id: I8285135adfea8c7c7c154034eda86123cacd4ae3
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1538134
Reviewed-by: Hannes Payer <[email protected]>
Commit-Queue: Michael Lippautz <[email protected]>
Cr-Commit-Position: refs/branch-heads/7.4@{v8#25}
Cr-Branched-From: 3e8a733-refs/heads/7.4.288@{#1}
Cr-Branched-From: d077f9b-refs/heads/master@{#60039}
billti pushed a commit to billti/v8 that referenced this pull request Apr 6, 2020
Revision: fb0a60e

BUG=chromium:1053604
NOTRY=true
NOPRESUBMIT=true
NOTREECHECKS=true
[email protected]

Change-Id: I40329f730fbfc5d578f607ecadc7853d4d5bd351
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2062406
Reviewed-by: Michael Stanton <[email protected]>
Reviewed-by: Michael Hablich <[email protected]>
Commit-Queue: Michael Hablich <[email protected]>
Cr-Commit-Position: refs/branch-heads/8.1@{v8#25}
Cr-Branched-From: a4dcd39-refs/heads/8.1.307@{#1}
Cr-Branched-From: f22c213-refs/heads/master@{#66031}
palmer-dabbelt pushed a commit to palmer-dabbelt/v8 that referenced this pull request Jul 19, 2020
Add support for cross compiling to riscv64
Kwizatz pushed a commit to AeonGames/v8 that referenced this pull request Sep 2, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com

Change-Id: I4528d07e1af279b5268f91f1b011bb1a074337b8
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2307459
Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/8.5@{v8#25}
Cr-Branched-From: a7f8bc4-refs/heads/8.5.210@{#1}
Cr-Branched-From: dd58472-refs/heads/master@{#68510}
ceejatec pushed a commit to couchbasedeps/v8-mirror that referenced this pull request Sep 24, 2020
Revision: ac3c4fc

BUG=chromium:1004365
LOG=N
NOTRY=true
NOPRESUBMIT=true
NOTREECHECKS=true
[email protected]

Change-Id: I68ff199d7d071134d4c9b5f8aace6af9745e8146
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1835955
Reviewed-by: Ulan Degenbaev <[email protected]>
Cr-Commit-Position: refs/branch-heads/7.8@{v8#25}
Cr-Branched-From: 73694fd-refs/heads/7.8.279@{#1}
Cr-Branched-From: 2314928-refs/heads/master@{#63555}
ceejatec pushed a commit to couchbasedeps/v8-mirror that referenced this pull request Sep 24, 2020
{JavaScriptFrame::GetParameters} allocates a new {FixedArray}, hence
all object references need to be handified to survive that allocation.

(cherry picked from commit fba03ab)

TBR=​[email protected]

Bug: chromium:1000635
Change-Id: Ib0978d317a1ac1bb88aede6e2b90260b9af68951
Originally-reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1787429
No-Try: true
No-Tree-Checks: true
No-Presubmit: true
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1833691
Reviewed-by: Clemens Backes <[email protected]>
Commit-Queue: Clemens Backes <[email protected]>
Cr-Commit-Position: refs/branch-heads/7.7@{v8#25}
Cr-Branched-From: 4035531-refs/heads/7.7.299@{#1}
Cr-Branched-From: 1320c91-refs/heads/master@{#62881}
ceejatec pushed a commit to couchbasedeps/v8-mirror that referenced this pull request Sep 24, 2020
(cherry picked from commit b8bf5e5)

No-Try: true
No-Presubmit: true
No-Treechecks: true
Bug: chromium:1016450
Change-Id: I4801b5ffb0ebea92067aa5de37e11a4e75dcd3c0
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1873692
Reviewed-by: Georg Neis <[email protected]>
Commit-Queue: Nico Hartmann <[email protected]>
Cr-Original-Commit-Position: refs/heads/master@{#64469}
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1883547
Reviewed-by: Jakob Gruber <[email protected]>
Cr-Commit-Position: refs/branch-heads/7.9@{v8#25}
Cr-Branched-From: be181e2-refs/heads/7.9.317@{#1}
Cr-Branched-From: 0d7889d-refs/heads/master@{#64307}
ceejatec pushed a commit to couchbasedeps/v8-mirror that referenced this pull request Sep 24, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com

Change-Id: Ia7248ad1e1191216db7a997e8e1c1ccf50e7a826
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2002375
Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/8.0@{v8#25}
Cr-Branched-From: 69827db-refs/heads/8.0.426@{v8#2}
Cr-Branched-From: 2fe1552-refs/heads/master@{#65318}
ceejatec pushed a commit to couchbasedeps/v8-mirror that referenced this pull request Sep 24, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com

Change-Id: I45dbf58cf2f3d48da4965dcaf0a865a8a6b4e3be
Reviewed-on: https://chromium-review.googlesource.com/c/1476300
Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/7.3@{v8#25}
Cr-Branched-From: 9df9418-refs/heads/7.3.492@{v8#2}
Cr-Branched-From: be213cf-refs/heads/master@{#59024}
ceejatec pushed a commit to couchbasedeps/v8-mirror that referenced this pull request Sep 24, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com

Cq-Include-Trybots: luci.chromium.try:linux_chromium_rel_ng
Change-Id: Ie737916fce0580415518c0a6dae07e1a60225cc9
Reviewed-on: https://chromium-review.googlesource.com/c/1309310
Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/7.1@{v8#25}
Cr-Branched-From: f70aaa8-refs/heads/7.1.302@{#1}
Cr-Branched-From: 1dbcc78-refs/heads/master@{#56462}
ceejatec pushed a commit to couchbasedeps/v8-mirror that referenced this pull request Sep 24, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com

Change-Id: I42e1fea70b21c716343686c431cf099d642a43b0
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1679141
Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/7.6@{v8#25}
Cr-Branched-From: 2cb2573-refs/heads/7.6.303@{#1}
Cr-Branched-From: 201c509-refs/heads/master@{#61902}
ceejatec pushed a commit to couchbasedeps/v8-mirror that referenced this pull request Sep 24, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com

Change-Id: Ib0064797142c32bd6900effc7f51e9c5180f4abe
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/1598869
Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/7.5@{v8#25}
Cr-Branched-From: 35b9bf5-refs/heads/7.5.288@{#1}
Cr-Branched-From: 912b391-refs/heads/master@{#60911}
ceejatec pushed a commit to couchbasedeps/v8-mirror that referenced this pull request Sep 24, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com

Change-Id: I8de7e3fd09678f588e1da5987af26f8c9c03188b
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2237957
Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/8.3@{v8#25}
Cr-Branched-From: 1668abd-refs/heads/8.3.110@{#1}
Cr-Branched-From: 04a7a68-refs/heads/master@{#66926}
ceejatec pushed a commit to couchbasedeps/v8-mirror that referenced this pull request Sep 24, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com

Change-Id: Ic448a94f4d5fde08570028a6f1dd6db6385ef425
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2235657
Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/8.4@{v8#25}
Cr-Branched-From: 88ed2e3-refs/heads/8.4.371@{#1}
Cr-Branched-From: 35f88bf-refs/heads/master@{#67773}
Kwizatz pushed a commit to AeonGames/v8 that referenced this pull request Oct 28, 2020
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com

Change-Id: I586fbf94affb3798d320546e8bf47e4f5c2d3216
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2466563
Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/8.6@{v8#25}
Cr-Branched-From: a64aed2-refs/heads/8.6.395@{#1}
Cr-Branched-From: a626bc0-refs/heads/master@{#69472}
Kwizatz pushed a commit to AeonGames/v8 that referenced this pull request Feb 6, 2021
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com

Change-Id: I7987e1d11b7a41262ed2fa0c1a23dc44c2f7b90a
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2618218
Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/8.8@{v8#25}
Cr-Branched-From: 2dbcdc1-refs/heads/8.8.278@{#1}
Cr-Branched-From: 366d30c-refs/heads/master@{#71094}
Kwizatz pushed a commit to AeonGames/v8 that referenced this pull request Apr 29, 2021
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com

Change-Id: If6e22c970209efe11e2a99b4b2162c067a828042
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2764844
Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/9.0@{v8#25}
Cr-Branched-From: bd0108b-refs/heads/9.0.257@{#1}
Cr-Branched-From: 349bcc6-refs/heads/master@{#73001}
Kwizatz pushed a commit to AeonGames/v8 that referenced this pull request Jul 25, 2021
TBR=v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com

Change-Id: If90b516f7e1d5e9cd2c1090cf6c2d2889075b1f9
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2837259
Reviewed-by: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/9.1@{v8#25}
Cr-Branched-From: 0e4ac64-refs/heads/9.1.269@{#1}
Cr-Branched-From: f565e72-refs/heads/master@{#73847}
salewski pushed a commit to salewski/v8 that referenced this pull request Aug 26, 2021
Revision: e165544

NOTRY=true
NOPRESUBMIT=true
NOTREECHECKS=true
[email protected]

Bug: chromium:1236978
Change-Id: I057cf235a794165eb5888032d4e21157b91fc60d
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3097268
Reviewed-by: Patrick Thier <[email protected]>
Cr-Commit-Position: refs/branch-heads/9.3@{v8#25}
Cr-Branched-From: 7744dce-refs/heads/9.3.345@{#1}
Cr-Branched-From: 4b6b4ca-refs/heads/master@{#75728}
BenPope pushed a commit to BenPope/v8 that referenced this pull request Dec 10, 2021
(cherry picked from commit 9c87415)

Change-Id: I582ff9ccd98be1dcdeb3c2ecc51fd11f3527fa4c
No-Try: true
No-Presubmit: true
No-Tree-Checks: true
Bug: chromium:1271456
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3289180
Reviewed-by: Andreas Haas <[email protected]>
Commit-Queue: Clemens Backes <[email protected]>
Cr-Commit-Position: refs/branch-heads/9.6@{v8#25}
Cr-Branched-From: 0b7bda0-refs/heads/9.6.180@{#1}
Cr-Branched-From: 41a5a24-refs/heads/main@{#77244}
Kwizatz pushed a commit to AeonGames/v8 that referenced this pull request Jun 27, 2022
Change-Id: I0aa4e436ae3b17433bda6c2594096535c99b221a
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3593352
Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/10.1@{v8#25}
Cr-Branched-From: b003970-refs/heads/10.1.124@{#1}
Cr-Branched-From: e62f556-refs/heads/main@{#79503}
cb-robot pushed a commit to couchbasedeps/v8-mirror that referenced this pull request Aug 24, 2022
Change-Id: I0415c5be6a0c3566b7487422f3f3086c3f31499a
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3306656
Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/9.7@{v8#25}
Cr-Branched-From: 49162da-refs/heads/9.7.106@{#1}
Cr-Branched-From: a7e9b8f-refs/heads/main@{#77674}
cb-robot pushed a commit to couchbasedeps/v8-mirror that referenced this pull request Aug 24, 2022
…, etc.

- Fix unaligned accesses by inserting a move.
- These ops seem to have the issue:

https://source.chromium.org/chromium/chromium/src/+/master:v8/src/compiler/backend/ia32/instruction-selector-ia32.cc;drc=fbc697b50b13df0c4078ad9c1fc6991f1f670e2e;l=2150

Bug: chromium:1142528
(cherry picked from commit 421ea59)

Tbr: [email protected]
Change-Id: Ifa050d9a6be4fe0b42f425a4488f25d228f8e831
No-Try: true
No-Presubmit: true
No-Tree-Checks: true
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/2513864
Reviewed-by: Zhi An Ng <[email protected]>
Commit-Queue: Zhi An Ng <[email protected]>
Cr-Commit-Position: refs/branch-heads/8.7@{v8#25}
Cr-Branched-From: 0d81cd7-refs/heads/8.7.220@{#1}
Cr-Branched-From: 942c2ef-refs/heads/master@{#70196}
patrik-lengweiler pushed a commit to hexagon-geo-surv/v8 that referenced this pull request May 31, 2023
If the phi moves are the same for all predecessors, the move optimizer
will merge them by picking an arbitrary move among them,
moving it to the phi's block, and eliminating the moves in the
predecessor blocks.
However, phi inputs may have different width, and this can result in a
mismatch between the source and destination representation.
Always emit gap moves based on the destination operand's
representation, to ensure that in this case the wider phi inputs are not
truncated.

R=​[email protected]
CC=​[email protected]

Bug: chromium:1407571
(cherry picked from commit 0bfae5d)

Change-Id: Ic920731ef572f26ff3f1b04aa7bd642a122dafe9
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4197676
Reviewed-by: Tobias Tebbi <[email protected]>
Commit-Queue: Thibaud Michaud <[email protected]>
Cr-Commit-Position: refs/branch-heads/11.0@{v8#25}
Cr-Branched-From: 06097c6-refs/heads/11.0.226@{#1}
Cr-Branched-From: 6bf3344-refs/heads/main@{#84857}
patrik-lengweiler pushed a commit to hexagon-geo-surv/v8 that referenced this pull request May 31, 2023
[email protected]

Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng
Change-Id: I2447a0934a74a72c6c3cd1cd3854615c1ab5ed3e
Reviewed-on: https://chromium-review.googlesource.com/590719
Reviewed-by: v8 autoroll <[email protected]>
Cr-Commit-Position: refs/branch-heads/6.1@{v8#25}
Cr-Branched-From: 1bf2e10-refs/heads/6.1.534@{#1}
Cr-Branched-From: e825c43-refs/heads/master@{#46746}
patrik-lengweiler pushed a commit to hexagon-geo-surv/v8 that referenced this pull request May 31, 2023
Windows uses a right-to-left function arguments evaluation order.
To support optimisation for this variant, we generate two more
profiles for x86 and x64 which use Windows to build the instrumented V8.

This change adds the GN config to build a instrumented V8 using Windows.
It is part of a multi-change CL and deployed in the following order:

1*) Add GN args for instrumented builds (https://crrev.com/c/4388886).
2) Add Windows builders (https://crrev.com/c/4388887)
3) Add Windows tracks to PGO builder (https://crrev.com/c/4388421)
4) Refactor profiling script (https://crrev.com/c/4388888)
5) Include r-l-profiles in V8 builds (https://crrev.com/c/4388577)

Additionally, this change sets a default for the v8-target-cpu parameter
of the profile_only.py script. We remove the parameter in the PGO
builder in 3) and rename it in 4).

This change needs to be cherry-picked to active 11.2+ release branches.

Bug: v8:13647
Change-Id: Ifd2e07ad71ea43eb18ee6f8f119680d17e1b1ef7
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4393854
Reviewed-by: Liviu Rau <[email protected]>
Reviewed-by: Lutz Vahl <[email protected]>
Commit-Queue: Alexander Schulze <[email protected]>
Cr-Commit-Position: refs/branch-heads/11.2@{v8#25}
Cr-Branched-From: 755511a-refs/heads/11.2.214@{#1}
Cr-Branched-From: e6b1cce-refs/heads/main@{#86014}
patrik-lengweiler pushed a commit to hexagon-geo-surv/v8 that referenced this pull request May 31, 2023
Bug: chromium:1356308
(cherry picked from commit ae32940)

Change-Id: I1e132e96325296d180488774ef183daa36dc22c7
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/3915224
Reviewed-by: Darius Mercadier <[email protected]>
Cr-Commit-Position: refs/branch-heads/10.6@{v8#25}
Cr-Branched-From: 41bc743-refs/heads/10.6.194@{#1}
Cr-Branched-From: d5f29b9-refs/heads/main@{#82548}
patrik-lengweiler pushed a commit to hexagon-geo-surv/v8 that referenced this pull request May 31, 2023
[email protected]

Cq-Include-Trybots: luci.chromium.try:linux_chromium_rel_ng
Change-Id: I85b901e2db4ab634e2f4c2b1d8537ff8e80dcd4a
Reviewed-on: https://chromium-review.googlesource.com/1018422
Reviewed-by: v8 autoroll <[email protected]>
Cr-Commit-Position: refs/branch-heads/6.7@{v8#25}
Cr-Branched-From: 8457e81-refs/heads/6.7.288@{v8#2}
Cr-Branched-From: e921be5-refs/heads/master@{#52547}
patrik-lengweiler pushed a commit to hexagon-geo-surv/v8 that referenced this pull request May 31, 2023
Change-Id: I8703afcf840a1d5c165b5b333cba3d123b485579
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4295210
Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/11.1@{v8#25}
Cr-Branched-From: c77793a-refs/heads/11.1.277@{#1}
Cr-Branched-From: 95b79bf-refs/heads/main@{#85479}
patrik-lengweiler pushed a commit to hexagon-geo-surv/v8 that referenced this pull request May 31, 2023
…ason

This is the V8 side change of crrev.com/c/876103.

Message:
It can happen that WebAssembly.compileStreaming gets aborted when we
are not allowed to execute JavaScript code, and therefore are also not
allowed to reject the promise returned by WebAssembly.compileStreaming.
This can happen e.g. when the Chrome tab gets refreshed, which aborts
all downloads.

With this CL we do not pass a reason to Abort if we are not allowed to
execute JavaScript code. On the V8 side we can check the reason passed
to Abort and do not reject the promise if there is no reason passed.

CC=​​​[email protected]
R=​[email protected]

NOTRY=true
NOPRESUBMIT=true
NOTREECHECKS=true

Bug: chromium:803838
Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng
Change-Id: Ie5cc85c72b60383e9221318c18a4e1812d230692
Reviewed-on: https://chromium-review.googlesource.com/876091
Reviewed-by: Ben Titzer <[email protected]>
Commit-Queue: Andreas Haas <[email protected]>
Cr-Original-Commit-Position: refs/heads/master@{#50749}(cherry picked from commit 56fe243)
Reviewed-on: https://chromium-review.googlesource.com/901182
Reviewed-by: Clemens Hammacher <[email protected]>
Cr-Commit-Position: refs/branch-heads/6.5@{v8#25}
Cr-Branched-From: 73c55f5-refs/heads/6.5.254@{#1}
Cr-Branched-From: 594a1a0-refs/heads/master@{#50664}
patrik-lengweiler pushed a commit to hexagon-geo-surv/v8 that referenced this pull request May 31, 2023
Change-Id: I50a54a31a2fe4161714ab96a18b912a18f0fef7d
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4232419
Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/10.9@{v8#25}
Cr-Branched-From: 8ade6bf-refs/heads/10.9.194@{#1}
Cr-Branched-From: 9ff2515-refs/heads/main@{#84164}
patrik-lengweiler pushed a commit to hexagon-geo-surv/v8 that referenced this pull request May 31, 2023
…rnative fix]

What happened:
- When rewriting in DoParseFunction, the relevant function scope is no longer in
the scope stack.
- The correct scope is given to the PatternRewriter.
- PatternRewriter called to Parser::BuildIteratorCloseForCompletion.
- BuildIteratorCloseForCompletion would just call NewTemporary (which creates
a new temporary in Parser's current scope) instead of using the scope passed to
it and calling NewTemporary on it.
- Normally this went unnoticed, since it doesn't matter that much where the
temporary is.
- But in the lazy arrow func case, the Parser's scope at that point was the
already-resolved outer scope, and a DCHECK detected this problem.

Kudos & thanks to verwaest@ for a debugging session :)

BUG=chromium:761831

NOTRY=true
NOPRESUBMIT=true
NOTREECHECKS=true

Change-Id: I1e8474ce927be0330f4ba4efc0fc08fdcc328809
Reviewed-on: https://chromium-review.googlesource.com/650297
Commit-Queue: Marja Hölttä <[email protected]>
Reviewed-by: Toon Verwaest <[email protected]>
Reviewed-by: Adam Klein <[email protected]>
Cr-Original-Commit-Position: refs/heads/master@{#47877}(cherry picked from commit 138fbdb)
Reviewed-on: https://chromium-review.googlesource.com/670599
Reviewed-by: Michael Achenbach <[email protected]>
Cr-Commit-Position: refs/branch-heads/6.2@{v8#25}
Cr-Branched-From: efa2ac4-refs/heads/6.2.414@{#1}
Cr-Branched-From: a861ebb-refs/heads/master@{#47693}
patrik-lengweiler pushed a commit to hexagon-geo-surv/v8 that referenced this pull request May 31, 2023
Bug: chromium:1426767
Change-Id: I7d94b3b8be0ebc0901a294c3c55c27b38c2f1c5d
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4556944
Commit-Queue: Michael Achenbach <[email protected]>
Bot-Commit: Rubber Stamper <[email protected]>
Cr-Commit-Position: refs/branch-heads/11.4@{v8#25}
Cr-Branched-From: 8a8a1e7-refs/heads/11.4.183@{#1}
Cr-Branched-From: 5483d8e-refs/heads/main@{#87241}
patrik-lengweiler pushed a commit to hexagon-geo-surv/v8 that referenced this pull request May 31, 2023
This reverts commit 6fc2187.

Reason for revert: to trigger another roll

Original change's description:
> Whitespace to trigger CI
>
> Bug: chromium:1426767
> Change-Id: I7d94b3b8be0ebc0901a294c3c55c27b38c2f1c5d
> Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4556944
> Commit-Queue: Michael Achenbach <[email protected]>
> Bot-Commit: Rubber Stamper <[email protected]>
> Cr-Commit-Position: refs/branch-heads/11.4@{v8#25}
> Cr-Branched-From: 8a8a1e7-refs/heads/11.4.183@{#1}
> Cr-Branched-From: 5483d8e-refs/heads/main@{#87241}

Bug: chromium:1426767
Change-Id: I2ad4b16834364540980091d97114b49fdb250dc3
No-Presubmit: true
No-Tree-Checks: true
No-Try: true
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4558608
Reviewed-by: Michael Achenbach <[email protected]>
Bot-Commit: Rubber Stamper <[email protected]>
Commit-Queue: Liviu Rau <[email protected]>
Cr-Commit-Position: refs/branch-heads/11.4@{v8#29}
Cr-Branched-From: 8a8a1e7-refs/heads/11.4.183@{#1}
Cr-Branched-From: 5483d8e-refs/heads/main@{#87241}
patrik-lengweiler pushed a commit to hexagon-geo-surv/v8 that referenced this pull request May 31, 2023
This is a reland of commit 6fc2187

Original change's description:
> Whitespace to trigger CI
>
> Bug: chromium:1426767
> Change-Id: I7d94b3b8be0ebc0901a294c3c55c27b38c2f1c5d
> Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4556944
> Commit-Queue: Michael Achenbach <[email protected]>
> Bot-Commit: Rubber Stamper <[email protected]>
> Cr-Commit-Position: refs/branch-heads/11.4@{v8#25}
> Cr-Branched-From: 8a8a1e7-refs/heads/11.4.183@{#1}
> Cr-Branched-From: 5483d8e-refs/heads/main@{#87241}

Bug: chromium:1426767
Change-Id: Ia152e332a5c5f235dce8e72d5a0a1f1f3560329e
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4565293
Reviewed-by: Michael Achenbach <[email protected]>
Bot-Commit: Rubber Stamper <[email protected]>
Commit-Queue: Liviu Rau <[email protected]>
Commit-Queue: Rubber Stamper <[email protected]>
Auto-Submit: Liviu Rau <[email protected]>
Cr-Commit-Position: refs/branch-heads/11.4@{v8#31}
Cr-Branched-From: 8a8a1e7-refs/heads/11.4.183@{#1}
Cr-Branched-From: 5483d8e-refs/heads/main@{#87241}
patrik-lengweiler pushed a commit to hexagon-geo-surv/v8 that referenced this pull request May 31, 2023
This reverts commit 8f0d686.

Reason for revert: to trigger roll pipeline

Original change's description:
> Reland "Whitespace to trigger CI"
>
> This is a reland of commit 6fc2187
>
> Original change's description:
> > Whitespace to trigger CI
> >
> > Bug: chromium:1426767
> > Change-Id: I7d94b3b8be0ebc0901a294c3c55c27b38c2f1c5d
> > Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4556944
> > Commit-Queue: Michael Achenbach <[email protected]>
> > Bot-Commit: Rubber Stamper <[email protected]>
> > Cr-Commit-Position: refs/branch-heads/11.4@{v8#25}
> > Cr-Branched-From: 8a8a1e7-refs/heads/11.4.183@{#1}
> > Cr-Branched-From: 5483d8e-refs/heads/main@{#87241}
>
> Bug: chromium:1426767
> Change-Id: Ia152e332a5c5f235dce8e72d5a0a1f1f3560329e
> Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4565293
> Reviewed-by: Michael Achenbach <[email protected]>
> Bot-Commit: Rubber Stamper <[email protected]>
> Commit-Queue: Liviu Rau <[email protected]>
> Commit-Queue: Rubber Stamper <[email protected]>
> Auto-Submit: Liviu Rau <[email protected]>
> Cr-Commit-Position: refs/branch-heads/11.4@{v8#31}
> Cr-Branched-From: 8a8a1e7-refs/heads/11.4.183@{#1}
> Cr-Branched-From: 5483d8e-refs/heads/main@{#87241}

Bug: chromium:1426767
Change-Id: Ia5dd57d46bc1b3f31b388050332baa97a69c37a6
No-Presubmit: true
No-Tree-Checks: true
No-Try: true
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4570564
Bot-Commit: Rubber Stamper <[email protected]>
Commit-Queue: Rubber Stamper <[email protected]>
Auto-Submit: Liviu Rau <[email protected]>
Cr-Commit-Position: refs/branch-heads/11.4@{v8#33}
Cr-Branched-From: 8a8a1e7-refs/heads/11.4.183@{#1}
Cr-Branched-From: 5483d8e-refs/heads/main@{#87241}
patrik-lengweiler pushed a commit to hexagon-geo-surv/v8 that referenced this pull request May 31, 2023
…soning is on.

Revision: a4353d1

BUG=chromium:798964
LOG=N
NOTRY=true
NOPRESUBMIT=true
NOTREECHECKS=true
[email protected]

Change-Id: I9f3fd1655b3d1467309888e7f26dd81845fcc80f
Reviewed-on: https://chromium-review.googlesource.com/973004
Reviewed-by: Benedikt Meurer <[email protected]>
Commit-Queue: Jaroslav Sevcik <[email protected]>
Cr-Commit-Position: refs/branch-heads/6.6@{v8#25}
Cr-Branched-From: d500271-refs/heads/6.6.346@{#1}
Cr-Branched-From: 265ef0b-refs/heads/master@{#51624}
patrik-lengweiler pushed a commit to hexagon-geo-surv/v8 that referenced this pull request May 31, 2023
…sing.

When skipping inner funcs is enabled, we also track variables for top level
funcs. Thus, we also declared the function name for the function scope, even
though it was the function scope for a function whose preparsing was
aborted. This lead into declaring the function name twice.

The fix is to declare the function name only in the success case.

The code was "wrong" before too, but this was never a problem, since variable
tracking and aborting preparsing were enabled for disjoint sets of
functions (aborting preparsing only for top-level, and variable tracking for
non-top-level).

BUG=v8:5516,chromium:774475
NOTRY=true
NOPRESUBMIT=true
NOTREECHECKS=true

Change-Id: Ie6c321cc834cd946e8843f73916fa7dd75e9cd09
Reviewed-on: https://chromium-review.googlesource.com/720920
Reviewed-by: Toon Verwaest <[email protected]>
Commit-Queue: Marja Hölttä <[email protected]>
Cr-Original-Commit-Position: refs/heads/master@{#48592}(cherry picked from commit d69159d)
Reviewed-on: https://chromium-review.googlesource.com/728000
Reviewed-by: Michael Hablich <[email protected]>
Cr-Commit-Position: refs/branch-heads/6.3@{v8#25}
Cr-Branched-From: 094a7c9-refs/heads/6.3.292@{#1}
Cr-Branched-From: 18b8fbb-refs/heads/master@{#48432}
patrik-lengweiler pushed a commit to hexagon-geo-surv/v8 that referenced this pull request May 31, 2023
[email protected]

Cq-Include-Trybots: master.tryserver.chromium.linux:linux_chromium_rel_ng
Change-Id: I0a4c477d44b65ded3837c6ab692fa762a2dd9227
Reviewed-on: https://chromium-review.googlesource.com/847423
Reviewed-by: v8 autoroll <[email protected]>
Cr-Commit-Position: refs/branch-heads/6.4@{v8#25}
Cr-Branched-From: 0407506-refs/heads/6.4.388@{#1}
Cr-Branched-From: a5fc4e0-refs/heads/master@{#49724}
patrik-lengweiler pushed a commit to hexagon-geo-surv/v8 that referenced this pull request Jun 26, 2023
(cherry picked from commit b04f0af)

Cq-Include-Trybots: luci.v8.try:v8_numfuzz_dbg
Cq-Include-Trybots: luci.v8.try:v8_numfuzz_rel
Cq-Include-Trybots: luci.v8.try:v8_numfuzz_tsan_rel
Cq-Include-Trybots: luci.v8.try:v8_linux_noi18n_rel
Cq-Include-Trybots: luci.v8.try:v8_linux64_msan_rel
Cq-Include-Trybots: luci.v8.try:v8_linux64_ubsan_rel
Bug: chromium:1448884
Change-Id: Ib3de10c84f82d838458b8dde8deab913161b4d59
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4568005
Reviewed-by: Liviu Rau <[email protected]>
Commit-Queue: Michael Achenbach <[email protected]>
Cr-Original-Commit-Position: refs/heads/main@{#87874}
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4645251
Auto-Submit: Michael Achenbach <[email protected]>
Commit-Queue: Liviu Rau <[email protected]>
Cr-Commit-Position: refs/branch-heads/11.5@{v8#25}
Cr-Branched-From: 0c4044b-refs/heads/11.5.150@{#1}
Cr-Branched-From: b71d303-refs/heads/main@{#87781}
patrik-lengweiler pushed a commit to hexagon-geo-surv/v8 that referenced this pull request Jul 28, 2023
Change-Id: Ic90ad9f1c456e92a4e4281a779892f547cf0eef4
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4726742
Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/11.6@{v8#25}
Cr-Branched-From: e29c028-refs/heads/11.6.189@{v8#3}
Cr-Branched-From: 95cbef2-refs/heads/main@{#88340}
reposynch Bot pushed a commit to hexagon-geo-surv/v8 that referenced this pull request Sep 3, 2023
Change-Id: Ic150e1c0dc7e22b952410b4600260fa7d60f14ff
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4828371
Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/11.7@{v8#25}
Cr-Branched-From: fe60869-refs/heads/11.7.439@{#1}
Cr-Branched-From: aeb4552-refs/heads/main@{#89415}
reposynch Bot pushed a commit to hexagon-geo-surv/v8 that referenced this pull request Nov 13, 2023
Change-Id: I001f390531f1abb1adef8ddfdcc3c463508a264a
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/4919190
Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/11.8@{v8#25}
Cr-Branched-From: 935bdbf-refs/heads/11.8.172@{#1}
Cr-Branched-From: b82a911-refs/heads/main@{#89779}
reposynch Bot pushed a commit to hexagon-geo-surv/v8 that referenced this pull request Dec 17, 2023
When we spawn graph builders for inlined function bodies, we must
share the outer graph builder's InstanceCache with them, because
that InstanceCache needs to see memory.grow instructions occurring
in inlinees.

Bug: chromium:1507663
(cherry picked from commit 53bae9f)

Change-Id: I04f6f36128d078d3720aa87f11bfe4660296959b
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5110201
Reviewed-by: Manos Koukoutos <[email protected]>
Commit-Queue: Jakob Kummerow <[email protected]>
Cr-Commit-Position: refs/branch-heads/12.1@{v8#25}
Cr-Branched-From: b74ef6f-refs/heads/12.1.285@{#1}
Cr-Branched-From: 32857fb-refs/heads/main@{#91313}
reposynch Bot pushed a commit to hexagon-geo-surv/v8 that referenced this pull request Dec 24, 2023
Change-Id: Ic7950315c51a0c83f8665ff6de469e73088a1676
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5131894
Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/12.0@{v8#25}
Cr-Branched-From: ed7b4ca-refs/heads/12.0.267@{#1}
Cr-Branched-From: 210e75b-refs/heads/main@{#90651}
hubot pushed a commit that referenced this pull request Feb 8, 2024
Change-Id: I03946670d7fb41bf0788e98c0dda50185a5aeede
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5274771
Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/12.2@{#25}
Cr-Branched-From: 6eb5a96-refs/heads/12.2.281@{#1}
Cr-Branched-From: 44cf56d-refs/heads/main@{#91934}
hubot pushed a commit that referenced this pull request Mar 25, 2024
Change-Id: Ic497a510cf5c36b0a77d3c32f46729d6ee2b7ce6
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5387590
Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/12.3@{#25}
Cr-Branched-From: a86e197-refs/heads/12.3.219@{#1}
Cr-Branched-From: 21869f7-refs/heads/main@{#92385}
hubot pushed a commit that referenced this pull request Apr 15, 2024
Change-Id: I41373e275aaaedf592a66e28aad956b4134415d7
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5453232
Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/12.4@{#25}
Cr-Branched-From: 309640d-refs/heads/12.4.254@{#1}
Cr-Branched-From: 5dc2470-refs/heads/main@{#92862}
hubot pushed a commit that referenced this pull request May 22, 2024
Change-Id: Ie4ec53186451ccb4f23c7512ffeadb54f73fee27
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5559884
Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/12.5@{#25}
Cr-Branched-From: 15b9756-refs/heads/12.5.227@{#1}
Cr-Branched-From: 497d857-refs/heads/main@{#93350}
hubot pushed a commit that referenced this pull request Jun 10, 2024
Change-Id: I6eb6d60f9ea687a4a199b840a5672dadaad125e9
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5617250
Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/12.6@{#25}
Cr-Branched-From: 3c9fa12-refs/heads/12.6.228@{#2}
Cr-Branched-From: 981bb15-refs/heads/main@{#93835}
hubot pushed a commit that referenced this pull request Jul 11, 2024
Change-Id: I4cccca251dae9619928f549ca4e3b26c72832b36
Reviewed-on: https://chromium-review.googlesource.com/c/v8/v8/+/5689947
Bot-Commit: v8-ci-autoroll-builder <v8-ci-autoroll-builder@chops-service-accounts.iam.gserviceaccount.com>
Cr-Commit-Position: refs/branch-heads/12.7@{#25}
Cr-Branched-From: 35cc908-refs/heads/12.7.224@{#1}
Cr-Branched-From: 6d60e67-refs/heads/main@{#94324}
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants